)]}'
{"id":"openstack%2Fshade~691476","triplet_id":"openstack%2Fshade~master~Id5424c420070b39b71e4163bbd5a471a2a94cb9e","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"Id5424c420070b39b71e4163bbd5a471a2a94cb9e","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-26 10:36:23.000000000","updated":"2019-10-28 06:13:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c48cb92fa33932ac86acc87d8d773b93437c64f7","_number":691476,"virtual_id_number":691476,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2019-10-28 06:13:02.000000000","_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-26 11:20:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2019-10-28 05:28:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-26 11:20:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 05:28:45.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"9b7e9ac24a4bfce3862cb5a35ceaa5a790c41ef5","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-26 10:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6601807575cb9d55ac549ad6699e2b1f6dd0c6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 11:20:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a49041cef644bba86b20e04432ca064 : SUCCESS in 1m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32ef03305bb942ce810d34840d6cb64e : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa06a23a411a49aa99dc39eb0995f35f : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dba01515f6c04c05bd44361f7704a95e : SUCCESS in 2m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39dfea6ec8b540c9ba7d9b025647e562 : SUCCESS in 2m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87aa740f5c904c5a934b05a763d56e6d : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4477301e5249df915797d0041fb082 : SUCCESS in 3m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a69d1115fa2b487199fd3020315a8770 : SUCCESS in 2m 54s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/475b30e0ce6a48729e033e1c9af9e362 : SUCCESS in 42m 31s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/82dd1bcf0fec4b1585e0e3b60be26237 : SUCCESS in 41m 53s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/3467b9920b10440fb0abe39b773b6b2f : SUCCESS in 3m 39s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/bd4c54f3331b49cb81a2282e00b887ec : FAILURE in 4m 53s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/7c77211dd1d4422e9dbf64dfcb544c12 : FAILURE in 5m 43s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/69c1f7b54d0c4657ab795a5598bc33d1 : SUCCESS in 39m 44s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/19638536b69d4e05a9cf320a996155ad : SUCCESS in 33m 19s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/a4fd43472299482da2d5cf4555059efb : SUCCESS in 35m 54s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ba11a94da0d1404e9dfe4ec04c801569 : SUCCESS in 42m 39s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/ecfa8e1d247b46a5a806032d223dbb19 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"74a50d7bfcf8f32282c41730c4e4f0eaa8ce959d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 05:27:53.000000000","message":"Patch Set 1:\n\nshade is a client library, thus in phase 2 - see http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html . Please WIP","accounts_in_message":[],"_revision_number":1},{"id":"4634e5d2766e60cbea2bbdd2df392a18f1ed4615","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 05:28:45.000000000","message":"Patch Set 1: Code-Review-1\n\nthis still runs py27 tests","accounts_in_message":[],"_revision_number":1},{"id":"ef131921e4f3eec1fb09fa32c87f4c985a3ab163","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 06:13:02.000000000","message":"Patch Set 1:\n\n\u003e this still runs py27 tests\n\nI know it, thanks,","accounts_in_message":[],"_revision_number":1},{"id":"c48cb92fa33932ac86acc87d8d773b93437c64f7","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 06:13:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d264cee56edd05703c5644946cd61f112716525","revisions":{"4d264cee56edd05703c5644946cd61f112716525":{"kind":"REWORK","_number":1,"created":"2019-10-26 10:36:23.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/76/691476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/76/691476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/76/691476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/76/691476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/76/691476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/76/691476/1"}}},"commit":{"parents":[{"commit":"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-26 10:36:08.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-26 10:36:08.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: Id5424c420070b39b71e4163bbd5a471a2a94cb9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4d264cee56edd05703c5644946cd61f112716525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4d264cee56edd05703c5644946cd61f112716525"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
