)]}'
{"id":"openstack%2Fshade~719020","triplet_id":"openstack%2Fshade~stable%2Fussuri~Ic7d659156da39d9f9311a9615ad6ce2222d5f3d6","project":"openstack/shade","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ic7d659156da39d9f9311a9615ad6ce2222d5f3d6","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-10 14:16:19.000000000","updated":"2020-05-07 14:29:21.000000000","submitted":"2020-05-07 14:23:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719020-1588861382883-452ece20","meta_rev_id":"f56db876a619d8cf7db1be411432533165acc634","_number":719020,"virtual_id_number":719020,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 14:23:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-07 13:24:16.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-07 13:24:16.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-04-27 09:07:31.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 09:07:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-07 13:24:16.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-07 14:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39306ad5d521675fde34d4941e73c6f24c755037","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-10 14:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d53e6e252461f93a9ae4030e793d5d337985d611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 14:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a385934f25d4b5d97bd9899d54b8e1e : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57520088dc0d46329540e8058a358d23 : FAILURE in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0267fe9b3af4cc5a0a88d6d140f05f0 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5afda33a2120475f9cf60897240ecde5 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81cface9a83e41e7aced903c21503dca : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b3a543bdb74320bd6bf77af61b70dc : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2cad5340fc0461883655faff173b37a : SUCCESS in 3m 48s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/cf2829104a6449cfa3d449ab7ba88fbb : FAILURE in 3m 38s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/27c0852a248e41668e167f19e5d123bc : SUCCESS in 3m 47s\n\nWarning:\n  Comments left for invalid file shade/openstackcloud.py\n  Comments left for invalid file shade/_utils.py\n  Comments left for invalid file shade/_legacy_clients.py\n  Comments left for invalid file shade/_heat/template_utils.py\n  Comments left for invalid file shade/task_manager.py\n  Comments left for invalid file shade/meta.py\n  Comments left for invalid file shade/exc.py\n  Comments left for invalid file shade/_heat/template_format.py\n  Comments left for invalid file shade/tests/fakes.py\n  Comments left for invalid file shade/tests/unit/test_meta.py","accounts_in_message":[],"_revision_number":1},{"id":"13e5d9479f24e4866c8287dc89acc129d610d42f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 09:07:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3b724f56d736299cca6c0c39b0f32b0f633a217c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:26:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73b130e9dd824646993ab1773aa97d5b : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa1d357d30e47b7bc9f6d15442548ac : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62ab6a1f642d45449535976f1d8e45cc : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfe989c54945445c8f5c4ae9236db01a : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b472b05708e4e93af2f80b847e564f8 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6006445086247daa709c533716ba7e6 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdc676af9952436bac2e8663b20db248 : SUCCESS in 4m 05s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/295401e442c24244b9a7ebb6aae34565 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2505c5963c7452287b1657c0b87c52a3c592f4e9","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-07 13:24:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d515a54dca17f0f8e67e5a000973c899f96dfc6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:24:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad36611d18f1aaed75d8c05c66a8f1e27d6f3aa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:23:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cb816fba5610f801806665fbc7d6920b65f45105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:23:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/725408e6896f429fa30d3070063322c2 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afcae4cdd10844e6a1ca471b48984866 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0047abe71c0d4a83a32cf7eb99cbeef8 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a6efa627b57488998fae607e492644b : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73473bd7609043a1ad018da2bc899b67 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8715d6cb34e346f3ae3ad6b1d71f3e03 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1524cfe2af4cacb1099780d43a1402 : SUCCESS in 4m 33s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/bfd79cff3caa437b9bfede598eac9081 : SUCCESS in 49m 43s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/cffc013b1931498380b06f2a00a1f166 : SUCCESS in 6m 07s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/723fe64f97694cb4971279fde217a77d : SUCCESS in 46m 56s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/61709e4ffeec4802817a6424a0bacaaf : SUCCESS in 47m 31s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/ab4a207315db4234a708ddeae0725fe0 : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3164e6f96786a562e41bf63359310b83c3e90e1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:29:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ac12236f224105a38e3a93602b5c57 : SUCCESS in 3m 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2a6b028ce82407e8512dba0dbdc32ad : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c0d842d3d855df71b93cc0c86624b452a0402e5","revisions":{"d9b5326fca479d44753b05084e1f32fd31eb1043":{"kind":"REWORK","_number":1,"created":"2020-04-10 14:16:19.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/20/719020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/20/719020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/20/719020/1"}}},"commit":{"parents":[{"commit":"cb78d06d7a92660eac9f7a0bf761b265f219e761","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/cb78d06d7a92660eac9f7a0bf761b265f219e761"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:17.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic7d659156da39d9f9311a9615ad6ce2222d5f3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/d9b5326fca479d44753b05084e1f32fd31eb1043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/d9b5326fca479d44753b05084e1f32fd31eb1043"}]},"branch":"refs/heads/stable/ussuri"},"5c0d842d3d855df71b93cc0c86624b452a0402e5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-27 09:07:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/719020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/20/719020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/20/719020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/20/719020/2"}}},"commit":{"parents":[{"commit":"3c6f3709459b870416cd271b2a4e0352fe112997","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3c6f3709459b870416cd271b2a4e0352fe112997"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-27 09:07:01.000000000","tz":60},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic7d659156da39d9f9311a9615ad6ce2222d5f3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/5c0d842d3d855df71b93cc0c86624b452a0402e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/5c0d842d3d855df71b93cc0c86624b452a0402e5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
