)]}'
{"id":"openstack%2Fshade~719022","triplet_id":"openstack%2Fshade~master~Iaceea4271459eeb0c5066b2a9304754691c16712","project":"openstack/shade","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"Iaceea4271459eeb0c5066b2a9304754691c16712","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-10 14:16:25.000000000","updated":"2020-04-17 16:58:36.000000000","submitted":"2020-04-17 16:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719022-1587142349844-61dee02a","meta_rev_id":"2e933e0569c075d8e08a1ebc1602bfe0c1c9c2a8","_number":719022,"virtual_id_number":719022,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 16:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-17 14:55:27.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-04-17 14:55:27.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 14:55:27.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-17 16:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c73f86a6c80b8432529e443c2d8a78b7cf73b32","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-10 14:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49338e9ad0f4226ce4318d16689020609ab4ca6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 14:18:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/shade on branch master.  The error was:\n\n  The project template \"openstack-python3-victoria-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-victoria-jobs\n        - publish-openstack-docs-pti\n        - publish-to-pypi\n        - release-notes-jobs-python3\n        - shade-functional-tips\n        - shade-tox-tips\n      check:\n        jobs:\n          - bifrost-integration-tinyipa:\n              voting: false\n          - bifrost-integration-tinyipa-opensuse-15:\n              voting: false\n          - shade-ansible-stable-2.5-functional-devstack:\n              voting: false\n          - shade-ansible-functional-devstack\n          - shade-functional-devstack\n          - shade-functional-devstack-magnum\n          - shade-functional-devstack-python3\n      gate:\n        jobs:\n          - shade-ansible-functional-devstack\n          - shade-functional-devstack\n          - shade-functional-devstack-python3\n\n  in \"openstack/shade/.zuul.yaml@master\", line 217, column 3","accounts_in_message":[],"_revision_number":1},{"id":"771592ea1d501ede59c997c7a48db06b4e3fafbb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 15:15:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"40cabd86b9b3f33bf319407ee02e83e5d8820bc4","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-17 14:55:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"423bd9395a08ef1921e1f4dbeb4286d8b8f924d9","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-17 14:55:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce9f29b00bb98201685706a41729286015b9eaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 15:55:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/149320eacf9647dfafca1b79c02ddeb7 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e29d30ad3f14835aefdcfc0728bc0c3 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea837b9f1ea143018cb9e074416a2d62 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93144ba8854a4f72bc9e557cff463a77 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/287ac33b8ee649b1be9278a11300aa0c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f86f254b086a4d70be68369a8577d674 : SUCCESS in 5m 01s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/e4d23b7d0d2f48d4987899f77dbae3bb : SUCCESS in 45m 30s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/f080d38b6f004562a8d12d13e624faca : SUCCESS in 6m 03s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/5385f4c3598e43f383f634e2985993f8 : SUCCESS in 21m 01s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/37959784b6ce474a9d5bf2087a22c688 : FAILURE in 5m 10s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/5eb402a35e9e485ab56918c1a9b36140 : SUCCESS in 41m 55s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/af26105fec5241c28e64de818453f0ec : SUCCESS in 36m 00s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/1f855593b16048fc8724b16e45ddefd1 : SUCCESS in 43m 12s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a428f9a00ee64a3d99702de6bd505e19 : POST_FAILURE in 52m 54s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/136a528814de4284b9ceca5e579725db : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":3},{"id":"dedd31177d259ff544c07d6be4bb72bcfbd51941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 15:55:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77f8b723eeec36f8cf7dae757f9793d60f4a69ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 16:52:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e024e3a9c2524814a4165fb2a9d50a58 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6108f33fa67a496e99a881f20d946376 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0698c93ee5134daeb00f93a9428ad48a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90c05206f9bf435981ad18bbc52fd08f : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/311a80739063494185ddcd0c0c5c88f4 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f0fe9cf6ef4c1687311fcef5e4d281 : SUCCESS in 4m 47s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/1fded273c2b64c4cb07be6066b584db7 : SUCCESS in 42m 59s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/1bc2c90f809a40e8b961aeac954ed126 : SUCCESS in 5m 23s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/4bce60f9fa9247c3b6faab45e30ce10d : SUCCESS in 37m 10s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/16840b43c1f047ddac57eb826fc1204c : SUCCESS in 46m 51s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/cc03b8b01ca04aa9ba07ce6f5948e396 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":3},{"id":"4560c3c192bf869db0f6fa1a263a0fcb85f09cb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 16:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2287b1d90bde960e7e2565003948c6f262d4e98f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 16:58:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad06e6697f3d47eb885b337a4b7883a7 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3aaaac9d16a7630a637e44be35cb5cf52faaa8dc","revisions":{"cf51f172685e8470c6cf935b21304a7ef0831aae":{"kind":"REWORK","_number":1,"created":"2020-04-10 14:16:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/22/719022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/719022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/719022/1"}}},"commit":{"parents":[{"commit":"3530e3f5aeac1335e78e7a1aedec14909024b0e1","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3530e3f5aeac1335e78e7a1aedec14909024b0e1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:22.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Iaceea4271459eeb0c5066b2a9304754691c16712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/cf51f172685e8470c6cf935b21304a7ef0831aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/cf51f172685e8470c6cf935b21304a7ef0831aae"}]},"branch":"refs/heads/master"},"f5605d393ff8b100f5199f8aeebe2ea832b91e39":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-10 15:15:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/719022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/719022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/719022/2"}}},"commit":{"parents":[{"commit":"acd5f5cdb955442b55a0ef6610e97b80b9431df6","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/acd5f5cdb955442b55a0ef6610e97b80b9431df6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:22.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 15:15:29.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Iaceea4271459eeb0c5066b2a9304754691c16712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f5605d393ff8b100f5199f8aeebe2ea832b91e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/f5605d393ff8b100f5199f8aeebe2ea832b91e39"}]},"branch":"refs/heads/master"},"3aaaac9d16a7630a637e44be35cb5cf52faaa8dc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-17 14:55:17.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/22/719022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/719022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/719022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/719022/3"}}},"commit":{"parents":[{"commit":"ccb2031788e95a0908cf3fa416708717840348c1","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/ccb2031788e95a0908cf3fa416708717840348c1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-10 14:16:22.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-17 14:55:17.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Iaceea4271459eeb0c5066b2a9304754691c16712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3aaaac9d16a7630a637e44be35cb5cf52faaa8dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3aaaac9d16a7630a637e44be35cb5cf52faaa8dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
