)]}'
{"id":"openstack%2Fshade~719034","triplet_id":"openstack%2Fshade~master~Ifdcb3ca27c50a182c06d9bf39c518a24ea0b3472","project":"openstack/shade","branch":"master","topic":"hacking","hashtags":[],"change_id":"Ifdcb3ca27c50a182c06d9bf39c518a24ea0b3472","subject":"Cap hacking for py2 support","status":"ABANDONED","created":"2020-04-10 15:14:53.000000000","updated":"2020-04-10 16:41:44.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"437b1821e06fcf1f37d88407a459fdd4a84881b1","_number":719034,"virtual_id_number":719034,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-10 16:31:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 15:44:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 15:44:52.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-10 16:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b605dc76391ffb3656864616a3e7105454d45dc8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 15:14:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d28b9eb079ed8cbbcc2b563734708ec01c9322a2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-10 15:21:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78c3cf802501e8da49b27d77e84607d18857edea","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 15:26:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"705c65b6be22d0b80ca18365b37ac43b9389619e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 15:40:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72c50c1005955665aa28e585746d5f5aeed38728","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-10 15:44:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"049ea721c29c24b319cbc9bb81b7c02c02e55c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 16:31:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d7c191090294bf2a4770480b0bf65a0 : SUCCESS in 2m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9efda3aac074c29b6fce89f8d86cbb0 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6ec00304f9466b972b837bb920bde0 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be7fb23a87cb4f01ab17d20e86eb78da : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f33ef01acb644909b8e948bb33775d8f : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82b59cef596c4d39973e4d57d93806bf : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/678f9a34f1ca4efab33a733899fbbd8c : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced1d2c735064a98b15ac51f5089d484 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63b9a14de7042d99b6b2f263237f7a6 : SUCCESS in 3m 11s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ba17107ce60045edbc0e2fecf4c57a92 : FAILURE in 32m 12s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/e96c3121536c4ac5ab80b2818961329f : SUCCESS in 44m 51s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/aef131f6b1f54d38ba9112b27590734d : FAILURE in 3m 12s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/11973dbcf45c44d2848cb7e353abb2db : SUCCESS in 3m 55s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/99bab0196ec84c859012ff045239a391 : SUCCESS in 15m 51s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/92d456999e024813aaa060bd970f344a : FAILURE in 3m 47s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/ce446f3c808a46dfbc64a069d5ac6be5 : SUCCESS in 35m 38s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/b0ebb9fd8486425189597c055c2836ac : SUCCESS in 34m 33s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/dc6a7a34a56a4d30b79a208d8da2b1e1 : SUCCESS in 45m 16s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d3a1a724d7e343a3b9260c7279ac24dc : SUCCESS in 39m 43s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/1cb9c69bd933476b928589094931bb6d : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":2},{"id":"95c3ec833a3eb4fd59dacb7b7f9c859b01ad0dd6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 16:41:44.000000000","message":"Abandoned\n\nLooks like this also needs changes that are included in https://review.opendev.org/#/c/718068/ (which I completely forgot I had done)\n\nThat also looks like it could use some work.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09035f66d4f88519b8c9c8f035c56477e04b7582","revisions":{"cbe73cfe6456a6f1b6e01592d6b26fbdd68cb4d2":{"kind":"REWORK","_number":1,"created":"2020-04-10 15:14:53.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/719034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/34/719034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/34/719034/1"}}},"commit":{"parents":[{"commit":"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 15:13:30.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 15:13:30.000000000","tz":-300},"subject":"Cap hacking for py2 support","message":"Cap hacking for py2 support\n\nThis caps the version of hacking used, and the linters that it ends up\npulling in, to still work with python 2.\n\nUpdates the way requirements are installed for the pep8 job to pull in\nall test-requirements.\n\nChange-Id: Ifdcb3ca27c50a182c06d9bf39c518a24ea0b3472\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/cbe73cfe6456a6f1b6e01592d6b26fbdd68cb4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/cbe73cfe6456a6f1b6e01592d6b26fbdd68cb4d2"}]},"branch":"refs/heads/master"},"09035f66d4f88519b8c9c8f035c56477e04b7582":{"kind":"REWORK","_number":2,"created":"2020-04-10 15:40:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/719034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/34/719034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/34/719034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/34/719034/2"}}},"commit":{"parents":[{"commit":"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 15:13:30.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 15:40:27.000000000","tz":-300},"subject":"Cap hacking for py2 support","message":"Cap hacking for py2 support\n\nThis caps the version of hacking used, and the linters that it ends up\npulling in, to still work with python 2.\n\nUpdates the way requirements are installed for the pep8 job to pull in\nall test-requirements.\n\nChange-Id: Ifdcb3ca27c50a182c06d9bf39c518a24ea0b3472\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/09035f66d4f88519b8c9c8f035c56477e04b7582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/09035f66d4f88519b8c9c8f035c56477e04b7582"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
