)]}'
{"id":"openstack%2Fshade~720746","triplet_id":"openstack%2Fshade~master~I97179030993c0b49ffe93aa28ffc672c6b859439","project":"openstack/shade","branch":"master","topic":"py38","hashtags":[],"change_id":"I97179030993c0b49ffe93aa28ffc672c6b859439","subject":"Declare py38 support","status":"MERGED","created":"2020-04-17 15:07:15.000000000","updated":"2020-04-27 09:56:38.000000000","submitted":"2020-04-27 09:52:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720746-1587981164380-dfe5b5ea","meta_rev_id":"e8d21953a90275f71eca7cfc663e2973eb0da585","_number":720746,"virtual_id_number":720746,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 09:52:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-17 17:02:05.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-04-27 09:03:10.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-04-27 09:03:10.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 17:02:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-27 09:03:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-27 09:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77bfa3a0ffe68a0354537868f0dd919449d51b9e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-17 15:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40a3f47ed06663e0014a034b4e5bba44ba04af83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 16:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b508270b7bb542dfb14d2939fb3d1806 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7feacaa3bf2b426fa936ef8731534fb9 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de44473c183c4936b6da267200bd88d1 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66a0eea0d8634c8287184bd2fba985e2 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b24b7cb635246feb8f77342413f7bbd : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c40db14cba46d0b88730f527ab12ee : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b09604e2ea346678bc02be944b03793 : SUCCESS in 6m 29s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/c43cb818342a4cd5bfaa45f78ab6aa8f : SUCCESS in 49m 12s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/988b5c0c9a704532b70baf0364b99ddb : SUCCESS in 7m 51s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/70552900ea1e463e9c2a4e60e3dda9cf : SUCCESS in 19m 01s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/e58c7fac3e784b1a985eed0902c4f636 : FAILURE in 7m 01s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/5dbaf05a67294f9ca38b5b558fff81df : SUCCESS in 45m 21s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/0acd1ffd1dee4d4dbef15584188af61b : SUCCESS in 44m 26s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/8255ecba4ed548fa9fdc77075f475f3d : SUCCESS in 41m 16s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9618dcd654034de497030a8cffc73fce : SUCCESS in 39m 28s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/de4ff82226cd4c1db19880645356548d : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9f11edaa80636bb1bb3968c072667b4dd885c153","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-17 17:02:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"793498a0ccd3d8d50d9b3083babd9a5df9d92f03","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 09:03:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fabfb1c7f99b46e02be244d4ed3f802f06ece41d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09a1bcdd4ac11f5e213bf93a793be7df9f1e9c54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:52:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fbaf4a8f85ce92bda252db43b31ce60254d51bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:52:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d2f8afc70b02439cbc9d72a5cb2d5f8a : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28aa76a42fc8471fa2a4fef49d1c5178 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71262efe69743958b304041100d632d : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db9d8871878444e593884e0c9e9515ce : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1915428a95d64ccfac0cb7b4525c3f3a : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe4312427562428ea30c6add42d698fd : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8477cf8ab24463baa63c31622e1ee98 : SUCCESS in 4m 45s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/8079674069ab441da39c030ef8dfe151 : SUCCESS in 44m 41s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6b3be4cd15d84baaa437a98edbdeb580 : SUCCESS in 6m 04s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/92063387c0ee4b1dbb69fec1e86a92b7 : SUCCESS in 34m 38s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/534bca438d094f50b6929263e460ea80 : SUCCESS in 39m 40s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/4773393223e8456d920f80d03f0df595 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c1e97ed56a6fa79943c16fccf100daaa52ccce23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:56:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269bb5c0d44645148978559f4810fce4 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4814b683f47d75035fb9840421e70c3cc821a0f4","revisions":{"4814b683f47d75035fb9840421e70c3cc821a0f4":{"kind":"REWORK","_number":1,"created":"2020-04-17 15:07:15.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/720746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/46/720746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/46/720746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/46/720746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/46/720746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/46/720746/1"}}},"commit":{"parents":[{"commit":"3aaaac9d16a7630a637e44be35cb5cf52faaa8dc","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3aaaac9d16a7630a637e44be35cb5cf52faaa8dc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-17 15:05:44.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-17 15:05:44.000000000","tz":-300},"subject":"Declare py38 support","message":"Declare py38 support\n\nPython 3.8 is one of the new required runtimes for the Victoria release.\nNow that we have test coverage for py38 and it is passing, we can add\n3.8 as one of the supported runtimes in the project metadata.\n\nChange-Id: I97179030993c0b49ffe93aa28ffc672c6b859439\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4814b683f47d75035fb9840421e70c3cc821a0f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/4814b683f47d75035fb9840421e70c3cc821a0f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
