)]}'
{"id":"openstack%2Fshade~722773","triplet_id":"openstack%2Fshade~master~I73ad76314acbd73e97a310f166dc82a965c6acee","project":"openstack/shade","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"I73ad76314acbd73e97a310f166dc82a965c6acee","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:27:11.000000000","updated":"2020-04-27 23:04:40.000000000","submitted":"2020-04-27 23:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722773-1588028607834-e63aa4fb","meta_rev_id":"a1323335e1b980b902809545292a717d662ce9de","_number":722773,"virtual_id_number":722773,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 23:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-27 22:12:36.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-04-27 08:28:31.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-04-27 22:12:36.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 08:28:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-27 22:12:36.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-27 23:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a43c084cdd6c9d753ed92515b93c1e6372475f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:27:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa6ed52c5d888c712af01355fa088f0791f49a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:39:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c7a154f142845feaf785ce69a7d29ed : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08e37ca1f9584f6998d0eba050fdc3af : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78886cebbc3e4ea39e7e96927c7ee53b : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21b730d4b4c54c3783c985f5e2fbfe45 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426a69b006e649caa9ee92e63f93dfe4 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f4d61cbcb724152b3a1c8df2a4b7c20 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7725b6cb8ef44a3092da8ea4e91f5e88 : SUCCESS in 8m 27s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/cecf5f5389c24d1eb6578cd2c8e94f40 : SUCCESS in 47m 40s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/0aa67eed6fc14e7a91a040e26bf932a2 : FAILURE in 5m 52s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/28fe46451eba4347af783b16275753c8 : SUCCESS in 37m 15s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/6a5c2c322ac14aff9281956af16fb077 : SUCCESS in 37m 01s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/95eb6dde20a94d9a8c992b728b44ac57 : SUCCESS in 42m 01s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/0aa23fc7e3d3444087a5c4147bc915b9 : SUCCESS in 49m 04s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/201d7798bdac48378fb8732af4221149 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1a1ee81af49ba54e7059c4f2d544da16a6b20e2b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 08:28:31.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bed835080eab0deee6eef048f56039c3b3a688d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:25:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7bd4dfed0e844e169b4caeb2cb45cbf7 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9307ab90a4ee42a69d17765e1e3e55a3 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d0aa8437de4c87973344b02b137249 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f86eb718f3f472dbf581fd331d4c0f1 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc3e6c2957b4433a14951b6da954efb : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7ce5b86b5c746fdb4deb031cf8d94b6 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c88fc5d6d7b4e4b9db137f862da5e9d : SUCCESS in 6m 15s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f4d42fa53540499a899fa565b9ac5076 : SUCCESS in 46m 14s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/8fa8a3d7c637440eaaaed4677ee32b85 : SUCCESS in 5m 12s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/d092a1cde04d407eb6a0a127fb47f3e4 : SUCCESS in 39m 28s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/c21e4dc4e06040798250e190626d435d : SUCCESS in 37m 51s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/82346ddf82094b1bb953ed98131cf50e : SUCCESS in 43m 09s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/14ca4aeafb48428d94589a1ec30b1ab1 : SUCCESS in 44m 09s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/cadae794019148dcb00814ffc7f73881 : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":1},{"id":"af73e36c4cc9269336132bfb9518cd10dbec3ca7","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-27 22:12:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54b38bddb89775a2dfc45f7214d0e68582ef89ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:12:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c4a04aaa766c581358f50cdbf7b61d6d4125e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:03:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0f45a015a81043b0976ac17cf51fa8c8 : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c993c6e47e254c91acbde9f116b698e6 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d323ae39e084a7e8651eaddd00e2a3f : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a4ea0d8357a4941b95e19fa370bbc3b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d90d140d2b4d42afa2e51bb85ad9a13d : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57ce9374d6ee462fa883bc542a06516c : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1295403bcf4ad58275076beed909dd : SUCCESS in 3m 57s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/ccfc801911994f64b175cdedfc6a17b2 : SUCCESS in 42m 43s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/eba3da495e3f422ea00ad87123d67722 : SUCCESS in 4m 09s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/1587848e8b9d4e63ade651cd30479f92 : SUCCESS in 43m 13s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/87b614a81d66425dacefcf70edaab709 : SUCCESS in 47m 15s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/cf8804dbfac54519af701f46999ff377 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ca98fb1297910feae05db1ba7a1ecd0d073ea0aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:03:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea17fcf51823d7a498f9d17228b0597142d8d682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:04:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cddc0639c341fc84cc20bc80e3cde2 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e966a96fa317e2ffc0931d0c4c2bdf3973d40b44","revisions":{"e966a96fa317e2ffc0931d0c4c2bdf3973d40b44":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:27:11.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/73/722773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/73/722773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/73/722773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/73/722773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/73/722773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/73/722773/1"}}},"commit":{"parents":[{"commit":"3aaaac9d16a7630a637e44be35cb5cf52faaa8dc","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/3aaaac9d16a7630a637e44be35cb5cf52faaa8dc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:21.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:21.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: I73ad76314acbd73e97a310f166dc82a965c6acee\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e966a96fa317e2ffc0931d0c4c2bdf3973d40b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e966a96fa317e2ffc0931d0c4c2bdf3973d40b44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
