)]}'
{"id":"openstack%2Fshade~723353","triplet_id":"openstack%2Fshade~stable%2Fussuri~Ia211e31723ede592f19a8df6c18e45699ca25788","project":"openstack/shade","branch":"stable/ussuri","hashtags":[],"change_id":"Ia211e31723ede592f19a8df6c18e45699ca25788","subject":"Break out py27 job templates","status":"MERGED","created":"2020-04-27 09:07:31.000000000","updated":"2020-05-07 14:21:33.000000000","submitted":"2020-05-07 14:17:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723353-1588861022090-39316be2","meta_rev_id":"4807c7cf2324877ce2a98b7e9bb2129e70e84006","_number":723353,"virtual_id_number":723353,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 14:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-07 13:24:05.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-07 13:24:05.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 13:24:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-07 14:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3707fbadb3f208207c177c89c8fb65227108272f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 09:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59ea83167d79e9fa6afdefa0b2181f6feca2dc03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6084f61d560e4a76b2584d4e51492163 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f608717291d4a4d906e2f22a91b3d4a : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb98252328b34bb393fa5bada4b089bd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a47ae632e614364bf87edb390d233d3 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f01ab408a2b487da89626cded636cec : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d578ea88f24babbb509762700694d8 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e187ca5d42a94e6da9d1dc73a69b55d1 : SUCCESS in 5m 10s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/cc99862690d443fb91a8034a3afb138d : SUCCESS in 6m 59s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/1b50d9ef0e0d476db49a10b5d01a9bd8 : SUCCESS in 17m 42s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/7ac71de0abb143c5a67d3ac42eab791c : FAILURE in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8de5eb665a79c37ab3d0acea0cb5215accbef209","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-07 13:24:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"093485c89db8f47084e1f77232dea1f8a2cd731f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:24:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"608f823b27a2001122fb9ddbd40ec1016d1bf35b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:17:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d27c5c26ba4aa498d08bfeda2abb2c : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5b2d37dace4501b621ba899d994751 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2a092da4fbf4e33a69212ee53320f51 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/062bea9e7b42460487f541333e2205db : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20fa1ad3ad5b415d82346a832198863f : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5502a6dfc8ab4fcf854578b353747727 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d050c50692a4ad59c0ad84e8b4e0e06 : SUCCESS in 4m 40s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/698d1d60ca544ec88139b5b4da76edd6 : SUCCESS in 45m 40s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/f9692f7ef2074ea99a1a66a1299226db : SUCCESS in 5m 28s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/746a2058c1e74112a8b0f12aea326e0a : SUCCESS in 35m 21s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/e862a48a023e4a50b7f67e6ba5cea05a : SUCCESS in 42m 30s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/d4da2afc923048b8a30cc3411fd8417d : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"92ca77c57aa01978cc513019b0fb92149611db76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:17:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4807c7cf2324877ce2a98b7e9bb2129e70e84006","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:21:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baedf5e566d245909f5191d38530a76d : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79f4cd4eb4b54078904a85012f44d98e : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c450221d216c849c87e0e18768019591489b0ab","revisions":{"0c450221d216c849c87e0e18768019591489b0ab":{"kind":"REWORK","_number":1,"created":"2020-04-27 09:07:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/723353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/53/723353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/53/723353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/53/723353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/53/723353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/53/723353/1"}}},"commit":{"parents":[{"commit":"e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e7c7f29ae9b9ec84ad05ba984cb62a15e4c33e7e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 11:26:01.000000000","tz":-300},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-27 09:05:49.000000000","tz":60},"subject":"Break out py27 job templates","message":"Break out py27 job templates\n\nTo allow non-py27 supporting repos to use shade job templates, this\nbreaks out the py27 and py3 jobs into separate templates. Relevant repos\ncan then continue to use the existing template names without picking up\nthe unsupported job runtimes they no longer need.\n\nRemove the py27-tips jobs - they just don\u0027t make any sense any more.\n\nChange-Id: Ia211e31723ede592f19a8df6c18e45699ca25788\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 3db89db8c611e9bc96ea5c7737d160f1700421fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0c450221d216c849c87e0e18768019591489b0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0c450221d216c849c87e0e18768019591489b0ab"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
