)]}'
{"id":"openstack%2Fshade~730722","triplet_id":"openstack%2Fshade~master~I60c0214bfa1bb6614f4e776c3d8a352cfcbfcb8b","project":"openstack/shade","branch":"master","topic":"publish-to-pypi","hashtags":[],"change_id":"I60c0214bfa1bb6614f4e776c3d8a352cfcbfcb8b","subject":"Remove publish-to-pypi template","status":"MERGED","created":"2020-05-26 07:51:44.000000000","updated":"2020-06-01 23:02:01.000000000","submitted":"2020-06-01 22:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730722-1591052354909-f06ad413","meta_rev_id":"e86d66d7d54954d3322e22d4267941cf3e4cc712","_number":730722,"virtual_id_number":730722,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 22:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-01 19:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 22:08:01.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 22:08:01.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 19:26:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 22:08:01.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-01 22:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9393037f773df75d07247a8fd181827c8d068524","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-26 07:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33b07ba5cb701390c80ff6955a1c5b6ed8aabdda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:35:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56b15e2f8d004cbeae0652ff541ca669 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ee77b8857b4e24a14c2e5183770a51 : FAILURE in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d6f4f6914934faa8e5efee792a2f373 : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6700470ee6b14825986bf5d4d9b6caf5 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee443860fe3549ba8b2b68bb978c27c9 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165748c40f874ab1945260f4d8789c01 : SUCCESS in 4m 26s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/eb7b3fc321d74c5a930de79dbfd22d4c : SUCCESS in 42m 07s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3c13e80bc86240f88eda7e143dbba036 : SUCCESS in 5m 38s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/637bd86e76094595a12b0a036cd724e8 : SUCCESS in 17m 21s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/0eecde8832e9477bb878cbfca22780d6 : FAILURE in 4m 49s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/e733727c78dc46ad898fdc5938b4ca3d : SUCCESS in 35m 23s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/baf3294ccb194d6294118bf1751d0cef : SUCCESS in 34m 12s\n- shade-functional-devstack https://zuul.opendev.org/t/openstack/build/c09d6f07bb1e4841a18444874ee53d67 : FAILURE in 29m 24s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d30e7a5ef4474dc9826cb57178c0bd8e : FAILURE in 30m 35s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/cfcfbf9dda5e4a73962e69b92fb55c61 : FAILURE in 31m 32s\n\nWarning:\n  Comments left for invalid file shade/tests/unit/test_image.py\n  Comments left for invalid file shade/_heat/utils.py\n  Comments left for invalid file shade/openstackcloud.py\n  Comments left for invalid file shade/tests/unit/test_stack.py\n  Comments left for invalid file shade/tests/unit/test_security_groups.py\n  Comments left for invalid file shade/tests/unit/test_object.py\n  Comments left for invalid file shade/_heat/event_utils.py","accounts_in_message":[],"_revision_number":1},{"id":"e9370a771f86d557cb5d48d5e6434eb3c740f964","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-01 18:50:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"efdc6e5250b4a1ae35185d9800406fc58e0df863","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 19:26:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e5a99b679e425eead3c9e6699baafacc225dbce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 19:45:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b20f362de3ea4d4095e16f2350ee66a1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4628750791b4d49ae98b30c97ef231d : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94e34d920f1a4d0ba31955d766f3546a : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c1e3acceb904d078936edf7061f4175 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a541505a216b4b55b4bf915e510ce2f0 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a68618bfe394e16ab6ee0ca855e647d : SUCCESS in 4m 24s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/7289445a36554b41a994d9159ec54996 : SUCCESS in 43m 41s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2497410a19764354b2be7b532310dc64 : SUCCESS in 6m 53s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/33534ae06f7c42c2aae91f51b14fd780 : SUCCESS in 17m 49s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/6105a7a95e4b438fb635d5abcf99fa17 : SUCCESS in 18m 26s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/a134549cb00844a5987f9d72ac8b7032 : SUCCESS in 33m 57s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/a3f374314b304954a44361cbee58c239 : SUCCESS in 47m 57s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5fd209d1a1464213a415492e97cef8ac : SUCCESS in 41m 48s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/0dc1743631fe4d0295be26c63a321d9b : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":2},{"id":"06600860426834d38a25024c00865976a9e8bea8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-01 22:08:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"074c7e6ec99cdb8dd24d3029a357e68d8b7b6c37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:08:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b6cf6cd1f0902352b34d0b5b55109915c842b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:59:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/206b19b9d49a4edaa07e27d604547eba : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ae461a877a4b7984d5138c32eed9bf : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42c2a6e07ef94ba5bee76a1841457d80 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cc939f27de846a1b8942c718c7a5b0b : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08790fac39be4c069ff5f2e84003c0e7 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469a6a898b2943feb49e2bc21b967cb9 : SUCCESS in 3m 44s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/12c94d57aec14fd0801261e709876cb8 : SUCCESS in 48m 15s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/bd141734441442aeae1d7d818e9733bd : SUCCESS in 4m 27s\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/f1f73ffb5f9b454ca827988648b24a51 : SUCCESS in 40m 02s\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/394c636a669f4efc96a9d246782c098e : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d8258d32b873560aeb18c5d0bd255a507729c264","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a1ba61a1759d0b0c3e1c04a3904859cf0f7932a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 23:02:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a007598b9b0446469f5d8899a69acc08 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c74ee2e74e0ba3cff3bee2172f96f213b69dd279","revisions":{"eb97384aa59b0011285951918c02a31cb3f9fa6f":{"kind":"REWORK","_number":1,"created":"2020-05-26 07:51:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/730722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/730722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/730722/1"}}},"commit":{"parents":[{"commit":"0d5c2b5249725ccc391e87356bed6df7b9c777e0","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0d5c2b5249725ccc391e87356bed6df7b9c777e0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-26 07:50:40.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-26 07:50:40.000000000","tz":120},"subject":"Remove publish-to-pypi template","message":"Remove publish-to-pypi template\n\nThis template needs to be in project-config since the jobs are not\nbranch aware. The template is already setup in project-config, so just\nremove the extra line here.\n\nChange-Id: I60c0214bfa1bb6614f4e776c3d8a352cfcbfcb8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/eb97384aa59b0011285951918c02a31cb3f9fa6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/eb97384aa59b0011285951918c02a31cb3f9fa6f"}]},"branch":"refs/heads/master"},"c74ee2e74e0ba3cff3bee2172f96f213b69dd279":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-01 18:50:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/730722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/22/730722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/22/730722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/22/730722/2"}}},"commit":{"parents":[{"commit":"08c07018dce264a8289aa0eeba3c834b772823d7","subject":"Fix pep8 and py27, functional jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/08c07018dce264a8289aa0eeba3c834b772823d7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-26 07:50:40.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-01 18:50:07.000000000","tz":0},"subject":"Remove publish-to-pypi template","message":"Remove publish-to-pypi template\n\nThis template needs to be in project-config since the jobs are not\nbranch aware. The template is already setup in project-config, so just\nremove the extra line here.\n\nChange-Id: I60c0214bfa1bb6614f4e776c3d8a352cfcbfcb8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/c74ee2e74e0ba3cff3bee2172f96f213b69dd279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/c74ee2e74e0ba3cff3bee2172f96f213b69dd279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
