)]}'
{"id":"openstack%2Fshade~755154","triplet_id":"openstack%2Fshade~master~Id78919ea93ca12abdc911d416fe8fbdbbf97264e","project":"openstack/shade","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"Id78919ea93ca12abdc911d416fe8fbdbbf97264e","subject":"Remove six","status":"ABANDONED","created":"2020-09-30 01:34:44.000000000","updated":"2020-12-22 11:48:39.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b525f7f9d07c064f8dc052543af79342a711d1b","_number":755154,"virtual_id_number":755154,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"date":"2020-10-08 00:45:53.000000000","_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-30 09:22:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-10-07 19:26:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 06:39:40.000000000","updated_by":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"reviewer":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"state":"REVIEWER"},{"updated":"2020-09-30 09:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 19:26:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5d111cc5d19c71ac2dd173f3ea0804eb8eee5180","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 01:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"385b164cbd33fd7e1f97ec47c4f5efb53784aeba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 03:14:37.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa872eb4311f438f8b5d28ac0c963c8e : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77f3b3d6cb014bcaaae81a9993f7909a : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b76ffe01c4b4f92a72e77bf2628cbab : FAILURE in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ed1780f94b144ce8660b286deafb7e4 : FAILURE in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb414a60ceee498c93293597f8247dbd : FAILURE in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77f837646ecf4ce5823922ded8e2651e : FAILURE in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6b2a210ef54e75b10c21f82ac97689 : FAILURE in 3m 17s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/622459a0d0124cb780e4f606621e49f3 : FAILURE in 38m 56s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d7f7072adaf24c5786e88de804f9ccf0 : FAILURE in 3m 58s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/d0c990e4142546e48fa93c9d6a2dc168 : SUCCESS in 23m 09s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/728868f7ac3b48f8b4bdc2cf936c0525 : SUCCESS in 25m 17s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/aed53d3b0b9d47158c18ae37c5c4cb89 : FAILURE in 36m 51s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/8349b3f9e2a74e0d8d694aa43866035c : FAILURE in 37m 10s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/70dea876241b4690bebd27390a531f16 : RETRY_LIMIT in 32m 55s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/06e2a08c4a12448797088bedb2544050 : FAILURE in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0db137adea9ec7971e4e2336332a4c754dd11100","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 05:53:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79231307101f8c8477f1b97878aed5ce8a040398","author":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"date":"2020-09-30 06:39:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69585d7f149157def9c828fbf72ce0e02092f277","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 06:53:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6a440d9454d15cac49bf9ab9b2138a39a93bac94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 09:22:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4380a0fe30e4a299ba5f57282569b8e : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3c3b535a8ea4b31929deb87dc10495f : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7effc5f3c54f4242ad828fda8ddc1be3 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2af4dac84f174a7b8c33d6b37e01f5d1 : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce5c05ed4a694cbb9f62898772e33023 : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19a8c82116354d4aade7f2652cd63173 : FAILURE in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa7e8bdf1914d87afdc9c3b4004ddc9 : FAILURE in 4m 44s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/0eea4a9726724acb8558bfe487a351e8 : FAILURE in 40m 25s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d650f8de49e34a61aae07983295c555e : FAILURE in 5m 30s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/ab735ad6d9884ed1941f387a76edea81 : SUCCESS in 23m 46s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f3a62090716a4127972eabc1770493b6 : SUCCESS in 21m 25s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/0af9861ed3fa41e7a57971846dcc8dcb : FAILURE in 44m 56s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/9bb6fe779f554526965a23d51b893cef : FAILURE in 36m 33s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/92a91894314144f1a1ab36542b7ac0ac : RETRY_LIMIT in 51m 44s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/5ee0f172e43e42d3975199b8cd5a0602 : FAILURE in 45m 12s","accounts_in_message":[],"_revision_number":3},{"id":"8c5320bc704f1c2b833b7b7dccb7e185c31bf6cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 19:25:56.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":3},{"id":"901bc40758ffde374a1a5f583613b96b4b7ba9e3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 19:26:13.000000000","message":"Patch Set 3: Code-Review-1\n\nThis repo still supports py2, so removal of six is not appropriate at this time.","accounts_in_message":[],"_revision_number":3},{"id":"764cdbad3701f27630ff47badb5a0c92726b8d34","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-08 00:45:53.000000000","message":"Patch Set 3:\n\n\u003e This repo still supports py2, so removal of six is not appropriate\n \u003e at this time.\n\nok， got it， thanks for review","accounts_in_message":[],"_revision_number":3},{"id":"8b525f7f9d07c064f8dc052543af79342a711d1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-22 11:48:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b3310ca86e15033db78097001f293a5fd36ac3b","revisions":{"7eb0e30f7b9f5ee5f768cd4cf2652391828dd8af":{"kind":"REWORK","_number":1,"created":"2020-09-30 01:34:44.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/54/755154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/54/755154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/54/755154/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:34:28.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:34:28.000000000","tz":480},"subject":"Remove six Replace the following items with Python 3 style code.","message":"Remove six\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.binary_type\n- six.string_types\n- six.add_metaclass\n- six.BytesIO\n\nChange-Id: Id78919ea93ca12abdc911d416fe8fbdbbf97264e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/7eb0e30f7b9f5ee5f768cd4cf2652391828dd8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/7eb0e30f7b9f5ee5f768cd4cf2652391828dd8af"}]},"branch":"refs/heads/master"},"28771952ef73b47eced4919b55f67932bab99289":{"kind":"REWORK","_number":2,"created":"2020-09-30 05:53:35.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/54/755154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/54/755154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/54/755154/2"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:34:28.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 05:53:23.000000000","tz":480},"subject":"Remove six Replace the following items with Python 3 style code.","message":"Remove six\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.binary_type\n- six.string_types\n- six.add_metaclass\n- six.BytesIO\n\nChange-Id: Id78919ea93ca12abdc911d416fe8fbdbbf97264e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/28771952ef73b47eced4919b55f67932bab99289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/28771952ef73b47eced4919b55f67932bab99289"}]},"branch":"refs/heads/master"},"6b3310ca86e15033db78097001f293a5fd36ac3b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-30 06:53:23.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/54/755154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/54/755154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/54/755154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/54/755154/3"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:34:28.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 06:53:12.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.binary_type\n- six.string_types\n- six.add_metaclass\n- six.BytesIO\n\nChange-Id: Id78919ea93ca12abdc911d416fe8fbdbbf97264e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/6b3310ca86e15033db78097001f293a5fd36ac3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/6b3310ca86e15033db78097001f293a5fd36ac3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
