)]}'
{"id":"openstack%2Fshade~768429","triplet_id":"openstack%2Fshade~master~I7138b309b1adc8686be994d0a5d4a0cbe0a3caa6","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"I7138b309b1adc8686be994d0a5d4a0cbe0a3caa6","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-12-24 06:57:00.000000000","updated":"2025-09-17 17:02:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4478f02961c27087b6c682ec45c2ea1f7cf279ff","_number":768429,"virtual_id_number":768429,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-24 08:56:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2020-12-29 01:02:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 08:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-29 01:02:26.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"fdd3c89dd29c9d0cbd444fabd0e59f9327dae8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-24 06:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12170857ca318b91e8bab2fa7b861706ed538840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 08:56:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea95e3cf986e4f9783c8b7c6ba715526 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f85602b78f41ec962517db6790cdf0 : SUCCESS in 2m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/870cd05102b24cd2b97ad1229748a406 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/231e5994d62d4be19e06e288e838f970 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86249aa96626438ca24547cb2b04489f : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a9cb4407ef4135ab0593de91ba97e2 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abadb910760c4b4f94eaf9f9cd09b1fd : SUCCESS in 3m 58s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/d7522884f3314ebe9ff640ded8905555 : SUCCESS in 38m 46s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2d17e9c8b650431388df5aed78617f83 : SUCCESS in 4m 02s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/3cdea85215b148f496e46bb2c58edf5b : FAILURE in 29m 13s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/d945f656b04b42a4bdab2cbcceb9cb93 : FAILURE in 27m 56s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/955850e793014128b8391f70f6bde84a : RETRY_LIMIT in 33m 31s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/5492bd2b85b44c5eadc06c97fc000e57 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0d06565a6dec94b26e9dc3d8f8a179648a3704b1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-29 01:02:26.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d79cd36bfcfb7962002e901d9095b68d18cc0f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:56:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c88af1fb4bbd45a48b143a67064de9a1 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d39a8be79c43b792048467af8fb8db : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/430973152ca04ef9a0306d66e646d73a : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e1e531096fa450c8bbd9c1957dc71a7 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97aa159938394130a611bae50004ed16 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d65c049d502464baf93428348535449 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76a901093cb145dd992d545c39b37f41 : SUCCESS in 3m 12s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/4aeec90680c14bc7bc558e52d6f8ffd8 : SUCCESS in 47m 59s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b8ae37ac7f0d45ba8f3ed8b7643728aa : SUCCESS in 4m 05s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/10c6de5c2063423eae216307a15eb363 : FAILURE in 36m 36s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/2b001838902c4f9fbe6360b2c8c2f55a : FAILURE in 35m 44s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c40e4998a6494ba8ab5d8208bb291386 : RETRY_LIMIT in 31m 00s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/8432f966dc13418291105afbe959a4af : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ab32949869917c09c0a1963d967fe28f35c6e080","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-31 00:30:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec00da0bc7e505a389656a3e79e9a1de783cdb70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 02:00:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/708d82d4bcbf4f468352de51ad527660 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61fe33d79854f59a3842d91aec5e654 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a883409fea98453ca0e195d2ed0c158c : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3e78c87af5b4ccb9c5193ed2b9d4c08 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d275828b6b4445098ac32d94abb6c01f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb616aa4af9496988c448b7bebe8530 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fd459db42824f44a0fac3f32ea23edc : SUCCESS in 3m 21s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/c0cfea05b6b7486a87c1b5b38d3a28ef : SUCCESS in 40m 53s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b9642d7a6bc546828d63aedb791698bb : SUCCESS in 4m 24s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/bce936497aaa465983f3cdfde818db39 : FAILURE in 31m 59s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/d29f33a45370470c9a30e31b3953ef9f : FAILURE in 29m 44s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/85484c670ab545c7a23cd2de7669ff41 : RETRY_LIMIT in 27m 58s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/24cac95c7d9148e09a7871ed94d6879c : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"775c62e8403e936d4b95e6314dd39120aa2eb44f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-01-02 01:40:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85b708de000c7dac9bf12a187d2e88cb441f1b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 03:20:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1233316e8a44501a2e4dfb60c1a467d : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b185eb01a7e40fcb7d563c827f1bd6f : SUCCESS in 2m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d2081b67243468ca7e1bc2c481661ac : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e33df09c8dc84d72be123bc4bcc64d15 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5507fdd18e2c435385ebd3d9fecc740f : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba923327f5842a5bdae50a65e784709 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf6485bfcd8344a3a1c05c0d0b863cad : SUCCESS in 2m 46s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f0429dd39a7e421faa0061abe5dd4a03 : SUCCESS in 38m 17s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/05aadabe15a44bd1b7eb4b7a7b74119f : SUCCESS in 3m 08s\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/fd07a70cc81142d99257041594000302 : FAILURE in 28m 56s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/a1241073d3254839bb641246184eb086 : FAILURE in 29m 40s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/f7b9fe3ab64d43bc8be2bc2f0c3dc57f : RETRY_LIMIT in 27m 48s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/0a3f2ddb056d48f98265d875c102c82d : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4478f02961c27087b6c682ec45c2ea1f7cf279ff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 17:02:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79f99529ef6cdee05fd6b074d1d73b1536b65481","revisions":{"79f99529ef6cdee05fd6b074d1d73b1536b65481":{"kind":"REWORK","_number":1,"created":"2020-12-24 06:57:00.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/29/768429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/29/768429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/768429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/768429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/29/768429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/29/768429/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 06:55:41.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 06:56:53.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I7138b309b1adc8686be994d0a5d4a0cbe0a3caa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/79f99529ef6cdee05fd6b074d1d73b1536b65481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/79f99529ef6cdee05fd6b074d1d73b1536b65481"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
