)]}'
{"id":"openstack%2Fshade~768886","triplet_id":"openstack%2Fshade~master~Ie6a9b7677e40497a5574253e94fbec8488739433","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"Ie6a9b7677e40497a5574253e94fbec8488739433","subject":"Add Victoria release info","status":"ABANDONED","created":"2021-01-03 00:46:58.000000000","updated":"2021-01-03 07:13:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad4a75127ab440c6bcf9f46e60965670f85dc482","_number":768886,"virtual_id_number":768886,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-03 01:59:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 00:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d074abe13305e4565ba1b5e64f11a41aae3abbc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 00:46:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52f38177c63b15c2627f5f384284e16d931df31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 00:54:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156beace06784c3081a7ee4f108a36c6 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6255a99ce54620b777986bb6e5b6b6 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f84c2a2a19e94f69856fbce2ae86461b : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"df6d1d0a54344db3b24fc010c1a4196856d3174e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 01:56:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b74288125dbd4e40b196b25f249e37272dc0dff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 01:59:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf2590c87e34b7281311ef4f0f3a3c5 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b5517f565d476fa1d0c26f1bebdfeb : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10b49bad791947e5a5b5ae55230f4106 : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ad4a75127ab440c6bcf9f46e60965670f85dc482","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-01-03 07:13:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"296c5c71e94481d46bc2c1ec489c4f0d7c2cdc09","revisions":{"755b97b3817e6d3f2eb66d67a01d3647fa7c7e3b":{"kind":"REWORK","_number":1,"created":"2021-01-03 00:46:58.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/86/768886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/86/768886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/86/768886/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 00:46:08.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 00:46:52.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: Ie6a9b7677e40497a5574253e94fbec8488739433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/755b97b3817e6d3f2eb66d67a01d3647fa7c7e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/755b97b3817e6d3f2eb66d67a01d3647fa7c7e3b"}]},"branch":"refs/heads/master"},"296c5c71e94481d46bc2c1ec489c4f0d7c2cdc09":{"kind":"REWORK","_number":2,"created":"2021-01-03 01:56:00.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/86/768886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/86/768886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/86/768886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/86/768886/2"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 00:46:08.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-01-03 01:55:19.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: Ie6a9b7677e40497a5574253e94fbec8488739433\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/296c5c71e94481d46bc2c1ec489c4f0d7c2cdc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/296c5c71e94481d46bc2c1ec489c4f0d7c2cdc09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
