)]}'
{"id":"openstack%2Fshade~768887","triplet_id":"openstack%2Fshade~master~I2bd0b0532924bc29a77f7ce4b1eba77099e73cdc","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"I2bd0b0532924bc29a77f7ce4b1eba77099e73cdc","subject":"Add Victoria release info","status":"ABANDONED","created":"2021-01-03 00:49:13.000000000","updated":"2021-05-05 01:27:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1fbbf46f2aa3b00708ab272e0fecc938be74a22a","_number":768887,"virtual_id_number":768887,"owner":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-03 07:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 07:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4051874b5e2cade1f63170687cfc8a48d1450acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-01-03 00:49:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adb05359e3c7e494279259331f07d04e05cb20d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-01-03 00:52:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"86957d489187d2b8a6c7ef3e9c074dbe27422794","tag":"autogenerated:gerrit:restore","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-01-03 07:15:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7d936d174980706248413347b64b5dcdaed6bb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 07:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115bc479d1c44d31b42c3f47e6377398 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbf884bd0404fd59c8a92461fdac034 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdeccf02d991433488d85d31577d0cae : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1fbbf46f2aa3b00708ab272e0fecc938be74a22a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2021-05-05 01:27:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e678e5d2f537deba70a57c3ddee97b21d33985a1","revisions":{"e678e5d2f537deba70a57c3ddee97b21d33985a1":{"kind":"REWORK","_number":1,"created":"2021-01-03 00:49:13.000000000","uploader":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"ref":"refs/changes/87/768887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/87/768887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/87/768887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/87/768887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/87/768887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/87/768887/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2021-01-03 00:48:43.000000000","tz":480},"committer":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2021-01-03 00:49:09.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I2bd0b0532924bc29a77f7ce4b1eba77099e73cdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e678e5d2f537deba70a57c3ddee97b21d33985a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/e678e5d2f537deba70a57c3ddee97b21d33985a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
