)]}'
{"id":"openstack%2Fshade~768900","triplet_id":"openstack%2Fshade~master~I7f35bcfe5d970ec150b998e46daf6ed1bd2bdf01","project":"openstack/shade","branch":"master","hashtags":[],"change_id":"I7f35bcfe5d970ec150b998e46daf6ed1bd2bdf01","subject":"Use unittest.mock instead of mock","status":"ABANDONED","created":"2021-01-03 01:51:16.000000000","updated":"2025-09-17 17:02:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c83ee12a06183a84904cba270ec60c996794c14b","_number":768900,"virtual_id_number":768900,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-03 03:28:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 03:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f06d0706dc702652522b82186359f2b162a28a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2021-01-03 01:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75039dda712e3f6687d3eb427d4a5e1ac9a93bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 03:28:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab751b01ab03472e87b37d16a173478a : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfac03121c5b46eba7e2cc4c477738c6 : SUCCESS in 2m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e5dd9218e2547c082c58d4915bbc3fd : FAILURE in 2m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0c61f45763d42c89244397411629cc5 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2814dcf3a30244c09f9c7523bb8a5928 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042d5da99cde48e9bf495cfdbc301ccf : SUCCESS in 2m 59s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/f5a94ffa36a64da9a90b437a2cfac9f8 : SUCCESS in 1h 29m 07s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dbe89aedadb54f12b1b0cf3f1b5a86a9 : SUCCESS in 3m 33s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/b21b5a41760349b2916a99a4d950457e : SUCCESS in 22m 19s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/f20664c78d2944fab12c3acbc41346cd : FAILURE in 2m 29s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/853f97f3599c4f73b5f7f39b2a391dc5 : FAILURE in 29m 05s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/5e956bc45aa741aaa9a0221bdf38afd7 : FAILURE in 28m 43s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/82a6a09bc9c640acbb7bbf144f350850 : RETRY_LIMIT in 33m 07s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/41ffcd2631584ed29ee8f8fa1a4acae3 : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c83ee12a06183a84904cba270ec60c996794c14b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 17:02:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c67812508914559b4a33a636864e0bcfb17c2ec9","revisions":{"c67812508914559b4a33a636864e0bcfb17c2ec9":{"kind":"REWORK","_number":1,"created":"2021-01-03 01:51:16.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/00/768900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/00/768900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/00/768900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/00/768900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/00/768900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/00/768900/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-01-03 01:50:00.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2021-01-03 01:50:00.000000000","tz":480},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I7f35bcfe5d970ec150b998e46daf6ed1bd2bdf01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/c67812508914559b4a33a636864e0bcfb17c2ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/c67812508914559b4a33a636864e0bcfb17c2ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
