)]}'
{"id":"openstack%2Fshade~790592","triplet_id":"openstack%2Fshade~master~Ia77bd52784a948fc60b6e28a2c3be6cf72562dc3","project":"openstack/shade","branch":"master","topic":"deprecation-warning","hashtags":[],"change_id":"Ia77bd52784a948fc60b6e28a2c3be6cf72562dc3","subject":"Replace getargspec with getfullargspec","status":"ABANDONED","created":"2021-05-11 08:32:39.000000000","updated":"2021-05-12 01:09:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af44ca30481c644fdf21aec0d18c9464721137f6","_number":790592,"virtual_id_number":790592,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-11 09:23:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 09:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb5be96de6d822966f38a33763483076f3570efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-11 08:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c09415d03f54676673bf608c5f933e9df4aea0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:23:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49c0cc8020c4ba2a392e3cfc629361d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76490b90c695456b89de3d6517a799cf : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f13016c1f724706af782eb451ca1e0d : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/531201c7c699472d81ab7098eff011e6 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a831fd7ecab24585ac9ef8b5b4e0e7ec : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f65bb6db4f48c6a258a2dfac8a9ab7 : SUCCESS in 4m 32s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/22aeb82222d64560a9284841c7f13941 : SUCCESS in 44m 40s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/799a562c628e4e34b906e3fd5a5f2b6c : SUCCESS in 5m 01s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/86f5e20700d347d79f077482576282f4 : SUCCESS in 33m 02s (non-voting)\n- bifrost-integration-tinyipa-opensuse-15 https://zuul.opendev.org/t/openstack/build/e088fd26141041b48922721850b475c1 : SUCCESS in 30m 37s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/2f960c377bb24e3eae2cd7382cfd2d38 : FAILURE in 38m 48s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/26f1729694bb4cf9bfa9e29f82ef6bd2 : FAILURE in 32m 18s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/68d897609f2a46958792e6b4c012a7a0 : SUCCESS in 40m 39s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/d969c3626666410ca0fdc11c5da5958b : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":1},{"id":"af44ca30481c644fdf21aec0d18c9464721137f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 01:09:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a37c20e89fd0116e8c2e2b43ad14e05ed761d871","revisions":{"a37c20e89fd0116e8c2e2b43ad14e05ed761d871":{"kind":"REWORK","_number":1,"created":"2021-05-11 08:32:39.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/92/790592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/92/790592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/92/790592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/92/790592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/92/790592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/92/790592/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 08:32:08.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 08:32:08.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: Ia77bd52784a948fc60b6e28a2c3be6cf72562dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a37c20e89fd0116e8c2e2b43ad14e05ed761d871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/a37c20e89fd0116e8c2e2b43ad14e05ed761d871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
