)]}'
{"id":"openstack%2Fshade~863851","triplet_id":"openstack%2Fshade~master~I312c7ee017ac26821734591f63b6879dcc2d3a19","project":"openstack/shade","branch":"master","topic":"drop-python-dev-from-bindep","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-09-17 17:03:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I312c7ee017ac26821734591f63b6879dcc2d3a19","subject":"Remove python-dev from bindep","status":"ABANDONED","created":"2022-11-07 10:08:14.000000000","updated":"2025-09-17 17:03:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d39853b6daced07a363983053a259a750a071325","_number":863851,"virtual_id_number":863851,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-07 10:46:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 10:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d717d7333d81799447517bbde099a72437c41d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-07 10:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd084ccd77840d8951d811e2a7dfa2f221534492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:46:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f2a137551ac4765a3bb90260941713e\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b0a4f9426424410935c911a4f8c6c8e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c849789f2aed4b1980ecf6fca06b7245 : SUCCESS in 2m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/069fa119c2224a818a8f3c6f9b1baaf9 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc920a96b0984918a4d75b2a21f136db : FAILURE in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f6daae435743d7b0f42cd33ff07ca1 : FAILURE in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c9ab5a505741f4aa765149d24b8b57 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/750021e935d54b93ae11e198f9bb3e4a : SUCCESS in 2m 56s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/99386cb2128644218776968d4d495585 : FAILURE in 32m 23s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fe0c394c13d04d9b8744b0a499b576bd : FAILURE in 3m 44s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/b39e21adac5f4ea4852153ec072dd092 : SUCCESS in 32m 08s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/5823ad55cd85453499c9229a29167793 : FAILURE in 29m 08s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/7ace4650f2e3475f822d384da8187dfc : FAILURE in 26m 09s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/93c6dd88169341358578aead2b4b2f2d : FAILURE in 33m 30s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/c91b01799dc64ef5be7aa61e693e6c5d : FAILURE in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d39853b6daced07a363983053a259a750a071325","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 17:03:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fecffbc7baed43dbc0ec2c263352783c7144a37","revisions":{"9fecffbc7baed43dbc0ec2c263352783c7144a37":{"kind":"REWORK","_number":1,"created":"2022-11-07 10:08:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/863851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/51/863851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/51/863851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/51/863851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/51/863851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/51/863851/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:08:10.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:08:10.000000000","tz":60},"subject":"Remove python-dev from bindep","message":"Remove python-dev from bindep\n\nIt is no longer supported by jammy and lead us to the following errors with the announce-release job.\n\n```\nNo package matching \u0027python-dev\u0027 is available\n```\n\nChange-Id: I312c7ee017ac26821734591f63b6879dcc2d3a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/9fecffbc7baed43dbc0ec2c263352783c7144a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/9fecffbc7baed43dbc0ec2c263352783c7144a37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
