)]}'
{"id":"openstack%2Fshade~937788","triplet_id":"openstack%2Fshade~master~I602ec0bb09181c498b0f78fcb33826bf65401041","project":"openstack/shade","branch":"master","topic":"zuul-config-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-12 09:21:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I602ec0bb09181c498b0f78fcb33826bf65401041","subject":"Drop missing bifrost job","status":"MERGED","created":"2024-12-16 15:25:57.000000000","updated":"2025-01-12 09:23:07.000000000","submitted":"2025-01-12 09:21:29.000000000","submitter":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937788-zuul-config-errors","meta_rev_id":"ab45849b86c22d7ef4d70b8fdd49d5ff450a7802","_number":937788,"virtual_id_number":937788,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-16 16:54:31.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-12 09:21:29.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-12-16 16:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-12-16 16:33:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 16:33:35.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-12-16 16:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-12 09:21:29.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"state":"REVIEWER"}],"messages":[{"id":"3ba490b5f06713e928dc4a098170fc447d76b773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-16 15:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d276409aa6ab985c6f05f0551656d2ccab08590","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-16 16:33:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f03672a230a1b0b001695b22c78dc0d6703faf76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:54:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17990f04b82d41cca2b635efe80261b2\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6536e20434045c28f66d974d6e4ddca : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a01b0424624f88a03dcf25115cd268 : RETRY_LIMIT in 2m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c64cc8a8b1ef4a469674bd865dff14cd : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f2ae9f1594740a7bf08eed498d54ba8 : FAILURE in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ac74adf28ae4b5dbc5b7bfd768b27fa : FAILURE in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235a2fbe23844621a34eeeea9eeefdd1 : RETRY_LIMIT in 2m 10s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/62d8be4d13954eab9250d7a2365716b3 : RETRY_LIMIT in 16m 14s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e99c130a4c044d1bb02a79ce73a40aac : FAILURE in 3m 11s\n- bifrost-integration-tinyipa https://zuul.opendev.org/t/openstack/build/246d27e2edad409e98d7bee993fba82d : SUCCESS in 24m 04s (non-voting)\n- shade-ansible-stable-2.5-functional-devstack https://zuul.opendev.org/t/openstack/build/437b0045b1374688aea9d7eae8385384 : RETRY_LIMIT in 20m 23s (non-voting)\n- shade-ansible-functional-devstack https://zuul.opendev.org/t/openstack/build/d5e9bf55577445bda19f4f000fff1233 : RETRY_LIMIT in 17m 33s\n- shade-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/efb47dddc18945cca8d9bb7892263ebd : RETRY_LIMIT in 24m 43s (non-voting)\n- shade-functional-devstack-python3 https://zuul.opendev.org/t/openstack/build/f4132ca8a3eb46b7998423bd6f3255b5 : RETRY_LIMIT in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3d4fda89ee8bf0227343ab0ab063289db57aee21","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2025-01-12 09:21:29.000000000","message":"Patch Set 1: Verified+2\n\nBypassing Zuul to merge this and get rid of the config error","accounts_in_message":[],"_revision_number":1},{"id":"8b66c8b29d813aa8676aaefe47b38be13fee7945","tag":"autogenerated:gerrit:merged","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2025-01-12 09:21:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab45849b86c22d7ef4d70b8fdd49d5ff450a7802","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 09:23:07.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93d6e258f2c5431ebf8a101525f27a2b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889c411fe73f4819a8f148d1c8254a63 : FAILURE in 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9dc7d04cc075d538e436f27084ba30dac94bc73","revisions":{"b9dc7d04cc075d538e436f27084ba30dac94bc73":{"kind":"REWORK","_number":1,"created":"2024-12-16 15:25:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/88/937788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/shade","ref":"refs/changes/88/937788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/shade refs/changes/88/937788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/shade refs/changes/88/937788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/shade refs/changes/88/937788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/shade refs/changes/88/937788/1"}}},"commit":{"parents":[{"commit":"0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3","subject":"Merge \"Remove publish-to-pypi template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/0ced9b5a7568dd8e4a33b6627f636639bcbbd8a3"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-16 15:24:28.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-16 15:25:50.000000000","tz":60},"subject":"Drop missing bifrost job","message":"Drop missing bifrost job\n\nThe bifrost-integration-tinyipa-opensuse-15 job no longer exists, drop\nit from the configuration in order to resolve the resulting error in\nzuul.\n\nChange-Id: I602ec0bb09181c498b0f78fcb33826bf65401041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b9dc7d04cc075d538e436f27084ba30dac94bc73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/shade/commit/b9dc7d04cc075d538e436f27084ba30dac94bc73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
