)]}'
{"id":"openstack%2Fskyline-apiserver~1007366","triplet_id":"openstack%2Fskyline-apiserver~master~Id31128e2fa867c499c062ce9f484d13555ce0059","project":"openstack/skyline-apiserver","branch":"master","attention_set":{},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-09-25 06:56:15.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}},"hashtags":[],"change_id":"Id31128e2fa867c499c062ce9f484d13555ce0059","subject":"Fix skyline-policy-manager generate-rule output format","status":"MERGED","created":"2026-09-25 03:57:56.000000000","updated":"2026-09-25 07:40:03.000000000","submitted":"2026-09-25 07:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007366","meta_rev_id":"2c9cd7af75878286f7443a0945913b0840504f99","_number":1007366,"virtual_id_number":1007366,"owner":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 07:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2026-09-25 06:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2026-09-25 06:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 04:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e81b95fac879336cd2d49121880086cb89e8948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-25 03:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9290ef4bc67a83cea8bb15d7461ad25d3aee5968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 04:35:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b227073821a74479b4a325b313db2da2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a598984a43d49368b1db42476ead8d8 : FAILURE in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7814a01c09e94157ba25811d3750d399 : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/632f6a8c8b03401c931e98bcbc3f491d : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/570c2a5045ad4415b33e8e2c5d207fb1 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164771b2707b4498b68511657500468d : SUCCESS in 8m 57s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/548fa933d3f240be9e03aea4137ef0aa : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d87e0ef2b9042f34a6969578b6f11d9de4fc372a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-25 05:52:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d9427542f453f59ff5d7214053b5bccceff6317e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:38:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f4420cdf3140c0b0dbbb7d68db8d28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff7bd6798874638873ede56cba0ba49 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f032a9d01dab40859f75a9ee0fee3015 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e904d6093d6d42af8a5697c56726afba : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ebb6ab2dce344428933291c50e3c16b : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9403847cf64b5eaa7d66c94b2ddfe8 : SUCCESS in 5m 56s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/141b50ddc0704013b536a9927e1a16db : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"427c54979cdb9cfa4b9ea65a3deb9e34c562e984","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-25 06:56:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fee3c307c8f577c4561ad9ec86815ff48eefcec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:56:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9509bb73fc7e1c176dcbcc3b013c11a17bfa3bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:39:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b4c356cd684d59bee80675d7fc8b78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e5c3165b4648bd8ffcd520a8bc0d08 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1330fd7841e643d591dde8846e250548 : SUCCESS in 2m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4fc90b4fa1b48988af01a60eb57c722 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8959e8c03d743da8834390eed8f2f26 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f89a23d23414481bb3f12a64e181e57 : SUCCESS in 8m 34s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/3fa294c39501445ab86413817629092a : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3bdda0221b8910dc6d5eef7c7862e0b0a12cef61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:39:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2c9cd7af75878286f7443a0945913b0840504f99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:40:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be38b13ab89e494baa7c5db92bed3546\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35dd697bdae84a1da17d554237f7a52d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c508d00d6a7f1f8f4cba8a0e0d8a39f4c1c8116b","revisions":{"e0d93213e4d161ccb14ecfed6411715b8e40db73":{"kind":"REWORK","_number":1,"created":"2026-09-25 03:57:56.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/66/1007366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/66/1007366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/1"}}},"commit":{"parents":[{"commit":"8650db5b8967677db269f32ba45a9e72cea6b78d","subject":"Merge \"Add Zaqar (messaging) configs and policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/8650db5b8967677db269f32ba45a9e72cea6b78d"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","date":"2026-09-25 03:57:43.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","date":"2026-09-25 03:57:43.000000000","tz":480},"subject":"Fix skyline-policy-manager generate-rule output format","message":"Fix skyline-policy-manager generate-rule output format\n\nEmit Operation imports and quoted rule fields so generated policy\nmanager modules match the checked-in style used by existing services.\n\nChange-Id: Id31128e2fa867c499c062ce9f484d13555ce0059\nSigned-off-by: Wu Wenxiang \u003cwu.wenxiang@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/e0d93213e4d161ccb14ecfed6411715b8e40db73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/e0d93213e4d161ccb14ecfed6411715b8e40db73"}]},"branch":"refs/heads/master"},"c508d00d6a7f1f8f4cba8a0e0d8a39f4c1c8116b":{"kind":"REWORK","_number":2,"created":"2026-09-25 05:52:04.000000000","uploader":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"ref":"refs/changes/66/1007366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/66/1007366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/66/1007366/2"}}},"commit":{"parents":[{"commit":"8650db5b8967677db269f32ba45a9e72cea6b78d","subject":"Merge \"Add Zaqar (messaging) configs and policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/8650db5b8967677db269f32ba45a9e72cea6b78d"}]}],"author":{"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","date":"2026-09-25 05:51:55.000000000","tz":480},"committer":{"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","date":"2026-09-25 05:51:55.000000000","tz":480},"subject":"Fix skyline-policy-manager generate-rule output format","message":"Fix skyline-policy-manager generate-rule output format\n\nEmit Operation imports and quoted rule fields so generated policy\nmanager modules match the checked-in style used by existing services.\n\nChange-Id: Id31128e2fa867c499c062ce9f484d13555ce0059\nSigned-off-by: Wu Wenxiang \u003cwu.wenxiang@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/c508d00d6a7f1f8f4cba8a0e0d8a39f4c1c8116b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/c508d00d6a7f1f8f4cba8a0e0d8a39f4c1c8116b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
