)]}'
{"id":"openstack%2Fskyline-apiserver~902413","triplet_id":"openstack%2Fskyline-apiserver~master~I6ef3b7b52634bca05de63033b0baed2163a65ce9","project":"openstack/skyline-apiserver","branch":"master","topic":"nginx-config","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2024-01-25 01:38:00.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"35814":{"account":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"last_update":"2024-01-25 02:22:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ef3b7b52634bca05de63033b0baed2163a65ce9","subject":"style: Uniform order of arguments, shorter lines","status":"MERGED","created":"2023-12-01 12:59:55.000000000","updated":"2024-01-25 02:23:38.000000000","submitted":"2024-01-25 02:22:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902413-nginx-config","meta_rev_id":"f64b0b03fc874ba48af10fd33e6f36c0a77b4dcb","_number":902413,"virtual_id_number":902413,"owner":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 02:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-01-25 01:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-01-25 01:38:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 13:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 13:34:26.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"aa0e9062238041ed3c4084787cb49188f6524f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"date":"2023-12-01 12:59:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5b9d60f1a28a1903a39cc23ea62da6834af4e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"date":"2023-12-01 13:02:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"30cc54f589460e303ba90dbaa28c532c6657e9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 13:31:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e479e9933af4ab390c910809057b55e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3c420a4041464ba32a740cf0cf5af9 : SUCCESS in 5m 04s\n- skyline-tox-py38 https://zuul.opendev.org/t/openstack/build/4c69d6967dda4eb380dbdb8fa9eed638 : SUCCESS in 3m 49s\n- skyline-tox-pep8-py38 https://zuul.opendev.org/t/openstack/build/e57ab5a745c247ed884a33419be2f94b : SUCCESS in 3m 43s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/76cf14c2146843e6984628949296ab02 : SUCCESS in 26m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e53b476297225060d75b2a5d3c7014a4a3d68b8e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-01-17 13:34:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b673b76af49cbfc3014b53a7a37482c0f5e5185a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"date":"2024-01-19 10:26:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d2e1f6c54d255ad9ba0b30bcfab6e3b436d57ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 11:07:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c8dae55cf74f40a288b57ba763609f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec15502d6824c7995c144f291aa0627 : SUCCESS in 5m 57s\n- skyline-tox-py38 https://zuul.opendev.org/t/openstack/build/878c1d1be48243d298b735d1d342df33 : SUCCESS in 5m 09s\n- skyline-tox-pep8-py38 https://zuul.opendev.org/t/openstack/build/faf5280f17e64efe9fcc8e2a04e5c6e3 : SUCCESS in 4m 34s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/9da94df80d8f40e48bd86c7d617ebd6c : SUCCESS in 33m 56s","accounts_in_message":[],"_revision_number":3},{"id":"20f525a9a950fa598fd3a94a2fa5188b47e7426c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-01-21 12:12:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f58c238906335019c85d1216585048a7bd8fe9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 12:13:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"729333a35d40dfca4ae6bb4eaff3b141a8bbcd91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 12:46:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b51ccc4d4515406c8c9009f686cb04b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341aa5d41eb043a1ac19f12b27545eae : FAILURE in 4m 54s\n- skyline-tox-py38 https://zuul.opendev.org/t/openstack/build/aacecc8558124384840284326c055594 : FAILURE in 2m 20s\n- skyline-tox-pep8-py38 https://zuul.opendev.org/t/openstack/build/b1b0f1c695fd420e9c0d1d5257f00c6b : FAILURE in 3m 08s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/0e5d206ec0fd497b972173a615149362 : FAILURE in 25m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b2c1097eadcf10faf8fd7db7e785b5c10b99dc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"date":"2024-01-24 12:05:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3202b83f2976f548352eaf15a4a201e12fd3e35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 13:18:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d8a389787c48d684e7f84596b68a17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb4f8527965498c8e6b72003fd6fd90 : SUCCESS in 5m 43s\n- skyline-tox-py38 https://zuul.opendev.org/t/openstack/build/7ed59dcf9d7448c4bc81c7762f02384b : SUCCESS in 3m 36s\n- skyline-tox-pep8-py38 https://zuul.opendev.org/t/openstack/build/a7db8d98d2814c2e90bcf80ce936fa89 : SUCCESS in 6m 06s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/a49581f5b1a4495596e0f785cc04af40 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":4},{"id":"df80ea326c914ec54e0acfbb7c46d7c44e6ee1c7","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-01-25 01:38:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b9f4de5d16222c94701cc522d1eac3eac436c00d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:38:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a4b4c43b95caeaa07b4fff37cc1accfb35b62788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 02:22:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64b82417bf34b67b0ce1134ac8f7ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34bd1647a84b41d8b175ff0bebb88c6b : SUCCESS in 5m 35s\n- skyline-tox-py38 https://zuul.opendev.org/t/openstack/build/e521c03291f24419a590159a19e8a812 : SUCCESS in 3m 38s\n- skyline-tox-pep8-py38 https://zuul.opendev.org/t/openstack/build/6f6090133daa4908809508b97a093838 : SUCCESS in 4m 01s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/789184341e5c4641b3d713d9506d2e05 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":4},{"id":"00b6fa89db7837d53e8e7027f7cd60faef079460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 02:22:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f64b0b03fc874ba48af10fd33e6f36c0a77b4dcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 02:23:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d32d019341a4ef09d7c5c6a68fbab7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e7ecc40406412ca827ddcd3bd8f53b : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"aef11398974abf8d02a4aa6a5801dca2dfdcddd6","revisions":{"acaf19eeb3f88fba8aea2f8a76a1d141610605ec":{"kind":"REWORK","_number":1,"created":"2023-12-01 12:59:55.000000000","uploader":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"ref":"refs/changes/13/902413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/13/902413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/1"}}},"commit":{"parents":[{"commit":"e0d390b83df92feda4ec57f255e8c6b311dfa5af","subject":"refactor: Prefer separate LABEL instructions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/e0d390b83df92feda4ec57f255e8c6b311dfa5af"}]}],"author":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 12:49:07.000000000","tz":60},"committer":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 12:50:50.000000000","tz":60},"subject":"style: Uniform order of arguments, tuned whitespace","message":"style: Uniform order of arguments, tuned whitespace\n\nExplicit line breaks make passed arguments clear.\n\nChange-Id: I6ef3b7b52634bca05de63033b0baed2163a65ce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/acaf19eeb3f88fba8aea2f8a76a1d141610605ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/acaf19eeb3f88fba8aea2f8a76a1d141610605ec"}]},"branch":"refs/heads/master"},"3e340618b6e5b078b972e58a88c398e504adfff9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-12-01 13:02:48.000000000","uploader":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"ref":"refs/changes/13/902413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/13/902413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/2"}}},"commit":{"parents":[{"commit":"4546837e14273e4dd285cfb9576a602d816975ad","subject":"refactor: Prefer separate LABEL instructions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4546837e14273e4dd285cfb9576a602d816975ad"}]}],"author":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 12:49:07.000000000","tz":60},"committer":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 13:02:00.000000000","tz":60},"subject":"style: Uniform order of arguments, shorter lines","message":"style: Uniform order of arguments, shorter lines\n\nExplicit line breaks make passed arguments clear.\n\nChange-Id: I6ef3b7b52634bca05de63033b0baed2163a65ce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/3e340618b6e5b078b972e58a88c398e504adfff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/3e340618b6e5b078b972e58a88c398e504adfff9"}]},"branch":"refs/heads/master"},"12fe4aeb6cabf14dc656d1660c24810f7c722f84":{"kind":"REWORK","_number":3,"created":"2024-01-19 10:26:42.000000000","uploader":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"ref":"refs/changes/13/902413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/13/902413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/3"}}},"commit":{"parents":[{"commit":"ce10b6691fc9fbdad4024e572a9352fb9a51671a","subject":"fix: user login failure due to no role assignment to its default project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/ce10b6691fc9fbdad4024e572a9352fb9a51671a"}]}],"author":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 12:49:07.000000000","tz":60},"committer":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2024-01-19 10:10:29.000000000","tz":60},"subject":"style: Uniform order of arguments, shorter lines","message":"style: Uniform order of arguments, shorter lines\n\nExplicit line breaks make passed arguments clear.\n\nChange-Id: I6ef3b7b52634bca05de63033b0baed2163a65ce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/12fe4aeb6cabf14dc656d1660c24810f7c722f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/12fe4aeb6cabf14dc656d1660c24810f7c722f84"}]},"branch":"refs/heads/master"},"aef11398974abf8d02a4aa6a5801dca2dfdcddd6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-24 12:05:59.000000000","uploader":{"_account_id":35814,"name":"Florian Haftmann","display_name":"fhaftmann","email":"florian.haftmann@informatik.tu-muenchen.de","username":"fhaftmann"},"ref":"refs/changes/13/902413/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/13/902413/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/13/902413/4"}}},"commit":{"parents":[{"commit":"b1a693d0a3a9ff1c44f0e1e56279fd238c0ea047","subject":"docs: add FAQ in README.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/b1a693d0a3a9ff1c44f0e1e56279fd238c0ea047"}]}],"author":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2023-12-01 12:49:07.000000000","tz":60},"committer":{"name":"Florian Haftmann","email":"flha@uhurutec.com","date":"2024-01-24 12:03:13.000000000","tz":60},"subject":"style: Uniform order of arguments, shorter lines","message":"style: Uniform order of arguments, shorter lines\n\nExplicit line breaks make passed arguments clear.\n\nChange-Id: I6ef3b7b52634bca05de63033b0baed2163a65ce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/aef11398974abf8d02a4aa6a5801dca2dfdcddd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/aef11398974abf8d02a4aa6a5801dca2dfdcddd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
