)]}'
{"id":"openstack%2Fskyline-apiserver~970991","triplet_id":"openstack%2Fskyline-apiserver~master~Id5a5d53583f79e1278f1d120f076d5225dac879b","project":"openstack/skyline-apiserver","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2025-12-17 12:15:26.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-12-17 11:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-17 12:15:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5a5d53583f79e1278f1d120f076d5225dac879b","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-12-14 22:27:03.000000000","updated":"2025-12-22 21:22:00.000000000","submitted":"2025-12-17 12:15:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970991","meta_rev_id":"9e0a3befc127decba9373210247215035326f830","_number":970991,"virtual_id_number":970991,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-17 12:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2025-12-17 11:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2025-12-17 11:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-14 22:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 08:08:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2025-12-15 08:08:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"1b4b8d646122e96c7c8cc98f1f29a82da2b4d1f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 22:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23daaf82999a9f8d218c759a183c8dc08be7e944","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 22:27:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"89f489df2830ad6695d4169c5488087a5e3b38ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 22:27:58.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"66e1b38acece8bf41fea756c1cfc61f7b29db8ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 22:33:39.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a65c3f8cf10a9707cd0e31b6a866a024fb8a7d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 22:56:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6da695ebacb54226a934e3ca33473abb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8417e2b696a4d9e95632e1ae6c5fc43 : FAILURE in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691965aef6354a4b8a81b65163de9369 : FAILURE in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5cb7e26dd574ba4bfce5842cbf9cf3e : FAILURE in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a8ef317f2b340fcabd06d02627841a4 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d7d33d602146f186e4d1eece1984ae : FAILURE in 6m 46s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/9e98692b4a3f48788585b3791c1bfaf2 : FAILURE in 16m 41s","accounts_in_message":[],"_revision_number":4},{"id":"8ff8b037623dfd0e56974fdf518addcdb2424de1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 23:09:55.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"40d12d5d165c4ae7615f4f3587b121f0b7fd2c34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-14 23:11:12.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6b3d094f6bc1f383a261dff621846b76ca7655d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 23:27:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d519a6471f9436ba4fa38d0bd6a807a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/33b08efa52e24a199f0f0e7a7a0826fa : FAILURE in 1m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5dafa14eb564a3ba946c2448191501c : FAILURE in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bace55fc4a84130b9cd4c6deb84c6a7 : FAILURE in 6m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73aada11672641c89b3fb5778d83aded : FAILURE in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7eee5bceed04b15889294d58822cb5b : FAILURE in 6m 11s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/66f4660267d047c9a7c1d9796eba99a1 : FAILURE in 11m 32s","accounts_in_message":[],"_revision_number":6},{"id":"efee8f35b1af1dc2e0a9f837b359b09d6a193d37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-15 00:12:05.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7320340ab59a1358564b4def8d7e3e6c631978f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 00:30:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bd53c2fd1b8480499e9fa14ce26438a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41e55f694ab9474dbd52cdb05da54f09 : FAILURE in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0399464a8c7c48a19944a09909ba454c : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d2d7621b82c4bcda8fb3d1970a2575b : FAILURE in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d7b69647c7d4259b27f11a7586e0424 : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a213ef03c1403abf1b727be44d99c4 : FAILURE in 3m 23s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/293118c38f1847a3a192bc0dcfa8706e : FAILURE in 17m 07s","accounts_in_message":[],"_revision_number":7},{"id":"4bb275f1ba30316786f1ffbc0e438bb02831faa2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-15 07:43:24.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"40dac23b9d9cd0424b46b1665fbf54fc561f1a57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 08:07:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd52aef8fa5746aea7f87d780b24fd3d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8856529231104a78815e23f4c115a721 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4fd4be0fa947959b30f6a499200015 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a91c6b69e1124d88a570a427d18dd9eb : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daf324be8e4647b2945898366dc8ae59 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f4283440b44538a41b68f5d13270a8 : SUCCESS in 6m 47s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/d3e03bca651240818b1210cd293ab99a : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":8},{"id":"90b85f30623f6088653d07235013ad123a8a72bb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-15 08:08:52.000000000","message":"Patch Set 8:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"e72798a0789c87c284a0a17f89a78dc10b81ad5b","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-12-17 11:35:17.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"076ad863cff6f07b7dcf9f621dd0ad5d604029c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 11:35:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"23e79cfadb74fdb53acd6860582fcb058d56e96b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 12:15:26.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a382f229ffb4af7b90b5540ecffe905\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dfe5ceac075242e5aac01b8dd6d0295b : SUCCESS in 1m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb295b606544d7e9b024ac71dac9f6f : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd2939751bd34bc29f089094bfefed7a : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34db6ae79e2e4732a6fd422b65aadf82 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6f0ace416241aa873a704877ba9ae0 : SUCCESS in 4m 14s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/0db3758517ff45aab0dbcad7b745bc6f : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":8},{"id":"5860932cfce24913ca71b9928df594ad0664732f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 12:15:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d6256207e1efdd94512c98c067b89c4b851cbb28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 12:16:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/555025f187c14e128a7ed6835292325a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafa4d82ae5a4db5ac45aa292f85f473 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":8},{"id":"52128f05e0ea5e999145bfc9a795f7a2bd819e07","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-12-22 13:05:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9e0a3befc127decba9373210247215035326f830","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-22 21:22:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f7c1f3a816cf4a47fc265bcd1cff21a21df489de","revisions":{"50ba7f058f1cab0ad614412c1ef03256e569bd5b":{"kind":"REWORK","_number":1,"created":"2025-12-14 22:27:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/1"}}},"commit":{"parents":[{"commit":"cd31aa512728564b7efa1c1f06a8f8c6d1d9f643","subject":"Replace glanceclient with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/cd31aa512728564b7efa1c1f06a8f8c6d1d9f643"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/50ba7f058f1cab0ad614412c1ef03256e569bd5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/50ba7f058f1cab0ad614412c1ef03256e569bd5b"}]},"branch":"refs/heads/master"},"9ce9481d26e741073fe1f8f02fe7e96604630571":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-14 22:27:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/2"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:40.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/9ce9481d26e741073fe1f8f02fe7e96604630571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/9ce9481d26e741073fe1f8f02fe7e96604630571"}]},"branch":"refs/heads/master","description":"Rebase"},"4ebc3cfb3097069e84c76265b91d2bb19c64fcb5":{"kind":"REWORK","_number":3,"created":"2025-12-14 22:27:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/3"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:58.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4ebc3cfb3097069e84c76265b91d2bb19c64fcb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4ebc3cfb3097069e84c76265b91d2bb19c64fcb5"}]},"branch":"refs/heads/master"},"0fe61707f56b82cbdf211536ee4419edd43bcbc8":{"kind":"REWORK","_number":4,"created":"2025-12-14 22:33:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/4"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:33:39.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/0fe61707f56b82cbdf211536ee4419edd43bcbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/0fe61707f56b82cbdf211536ee4419edd43bcbc8"}]},"branch":"refs/heads/master"},"41f232c37eb2754113034a119784640534dc7ba7":{"kind":"REWORK","_number":5,"created":"2025-12-14 23:09:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/5"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 23:09:55.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/41f232c37eb2754113034a119784640534dc7ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/41f232c37eb2754113034a119784640534dc7ba7"}]},"branch":"refs/heads/master"},"815b33614b9113bd0c71d021cab1a1ed42920792":{"kind":"REWORK","_number":6,"created":"2025-12-14 23:11:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/6"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 23:11:12.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/815b33614b9113bd0c71d021cab1a1ed42920792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/815b33614b9113bd0c71d021cab1a1ed42920792"}]},"branch":"refs/heads/master"},"2a3aed5426599712aaa3f7d87042b5dddf7dc919":{"kind":"REWORK","_number":7,"created":"2025-12-15 00:12:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/7"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-15 00:12:05.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/2a3aed5426599712aaa3f7d87042b5dddf7dc919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/2a3aed5426599712aaa3f7d87042b5dddf7dc919"}]},"branch":"refs/heads/master"},"f7c1f3a816cf4a47fc265bcd1cff21a21df489de":{"kind":"REWORK","_number":8,"created":"2025-12-15 07:43:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/970991/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/91/970991/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/91/970991/8"}}},"commit":{"parents":[{"commit":"4e1dba114ea967e05537ef0d168a69ec10a7151d","subject":"Drop support for Python 3.8 and 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/4e1dba114ea967e05537ef0d168a69ec10a7151d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-14 22:27:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-15 07:43:24.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id5a5d53583f79e1278f1d120f076d5225dac879b\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/f7c1f3a816cf4a47fc265bcd1cff21a21df489de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/f7c1f3a816cf4a47fc265bcd1cff21a21df489de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
