)]}'
{"id":"openstack%2Fskyline-apiserver~975921","triplet_id":"openstack%2Fskyline-apiserver~master~I628f4b6045f3368e4cc9af5599d139c2a51b3720","project":"openstack/skyline-apiserver","branch":"master","topic":"remove-glanceclient-typo","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-10 11:35:35.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-02-10 09:02:57.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I628f4b6045f3368e4cc9af5599d139c2a51b3720","subject":"Fix typo in list_images()","status":"MERGED","created":"2026-02-06 15:41:14.000000000","updated":"2026-02-10 11:36:27.000000000","submitted":"2026-02-10 11:35:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975921-remove-glanceclient-typo","meta_rev_id":"431c0f43ce8a940d2929b97b30cf97b1ff34d709","_number":975921,"virtual_id_number":975921,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-10 11:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2026-02-07 03:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":2,"date":"2026-02-10 09:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-08 21:31:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2026-02-10 09:03:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 16:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-07 03:21:52.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2026-02-07 03:23:10.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2026-02-08 21:31:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"df49dc3dfad9f668dbe8b7edc90ff9bbfe32e30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-06 15:41:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e909b6bcb70a285debf3dc860acf9016a56342b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 16:13:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c61a849e6d14e9ea1f742943759a477\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d9f9946d244bb590540532b92e486f : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b2622702604c91ba1cc960fc8b3f5e : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06c8942af8524ba78d6c2b2149a6e8e2 : FAILURE in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d27e5aa18f4edc937abaeeeca98b1e : SUCCESS in 5m 48s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/7c7e8fd15a1942e4a6cda412c4a0770e : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7d9a46566e7a7af5ccedc26cbfe1538d7454c58f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-06 17:19:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b8de8f0c3f9cdd71a9e385ae21b74d2311a82ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 18:00:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd2f7cf19fb4fcf97c7c6e1e9ff8aeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee8da19b84d49f9bd2c180fbd12c7db : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8240d76a5c424cd3ab49da56cf840e31 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76dab517cb934cf387d928eaa4b75ac6 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305956fef08c4d619bb84a731396a32a : SUCCESS in 5m 58s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/f90568573e544696a2085c4e16720a03 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c3475eff63736de8fc06c62ef3370164379028c5","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2026-02-07 03:21:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aab0abe7a0a4774a127f4a09d25e7de2f307360f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-08 21:31:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a83323a0b6ee478faa70e8fe3b8cdbba40f8ad6","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-02-10 09:02:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8d9516ddf2a6a55ebace2923c3e6d78002ee8c7","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-02-10 09:03:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e77457dec304ace48c357c736d29f3cf4aa9354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 09:03:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e4c42e46f30c81b4abf01d32b659737137a319c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 10:17:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b326b77c41e04465a840100f20623301\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2c13d94dd347f19c482874a9cd321c : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39ae7fcb0da347b9910e09901b89f4e0 : FAILURE in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ac1b25da73b4958b0d8094a8f0f3cae : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6b8dd03fe84d688ada9f9f1fe3a764 : SUCCESS in 7m 58s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/ca100ad7a99e4196a4308efea7dd09f8 : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"600d07b1e561c29fe9456a43508471189c6db0a7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 10:25:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f56f5568efb8093e5c5cb06b6040be254ae3e314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 10:44:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aed606e7d9146c8b95e6dbec12b599a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21aaec6237b0472d91d2d741a6cb39c0 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97394bdcf3ab4c62a16f2afd4bd3b7e5 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4670a95d5c24b8bb4b834c413a13db3 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612f6a5359244fea94e5d22caabd30de : SUCCESS in 7m 17s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/edf17089a4dc45a9b37de2c5213a77e7 : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"31f77862c58c345c293b71c7ea695ccad3c1c41f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 10:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11c3f17bdb69930d478dfe03f3ea16d0f54c3964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:35:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a967e09bf65e4b46aaadd9a19e20c711\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984422afed0342dfa4137c0bfd5eca48 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5363e4fe5574661a626f1293d1b65a3 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1247113037d04c779179cd7a286c5f76 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20bb63384cfe4946ac3c8ef6f0f9f5d6 : SUCCESS in 7m 18s\n- skyline-functional-devstack https://zuul.opendev.org/t/openstack/build/62780af1791c4d1791528928a19d4b92 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4fa8171e1ff4f7a979b3e6ba01c348debe9f9f58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:35:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"431c0f43ce8a940d2929b97b30cf97b1ff34d709","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:36:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1238956597664e74952baad998060adc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e85f2461184f0aaa862df7b32a8379 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3f999a4224a25c6c1d4e8425db082fff836a583","revisions":{"e3f999a4224a25c6c1d4e8425db082fff836a583":{"kind":"REWORK","_number":1,"created":"2026-02-06 15:41:14.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/21/975921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-apiserver","ref":"refs/changes/21/975921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/21/975921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/21/975921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-apiserver refs/changes/21/975921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-apiserver refs/changes/21/975921/1"}}},"commit":{"parents":[{"commit":"38b837416285716e3dd04379c13baeaa0fa139c8","subject":"Merge \"Remove ironic_inspector.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/38b837416285716e3dd04379c13baeaa0fa139c8"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-02-06 15:39:15.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-02-06 15:41:07.000000000","tz":60},"subject":"Fix typo in list_images()","message":"Fix typo in list_images()\n\nThis was introduced in cd31aa512728564b7efa1c1f06a8f8c6d1d9f643 .\n\nThanks to Lennart Nordgreen for reporting this issue.\n\nChange-Id: I628f4b6045f3368e4cc9af5599d139c2a51b3720\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/e3f999a4224a25c6c1d4e8425db082fff836a583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-apiserver/commit/e3f999a4224a25c6c1d4e8425db082fff836a583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
