)]}'
{"id":"openstack%2Fskyline-console~1005259","triplet_id":"openstack%2Fskyline-console~stable%2F2026.2~I7949a54e1ae6a3260ceb3ca0b383cb6700c1bfe9","project":"openstack/skyline-console","branch":"stable/2026.2","topic":"create-2026.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-09-21 12:36:03.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-09-18 23:07:15.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7949a54e1ae6a3260ceb3ca0b383cb6700c1bfe9","subject":"Update .gitreview for stable/2026.2","status":"MERGED","created":"2026-09-11 14:16:25.000000000","updated":"2026-09-21 12:37:04.000000000","submitted":"2026-09-21 12:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005259-create-2026.2","meta_rev_id":"da9bf1f48e9796bf6e99c05595ddf5ab6a369256","_number":1005259,"virtual_id_number":1005259,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 12:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":2,"date":"2026-09-18 23:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-14 14:00:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":1,"date":"2026-09-18 23:07:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 14:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 14:00:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-14 14:00:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-09-18 16:36:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2026-09-21 09:44:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"20b5602885ad5e44a50c93466e660f981bdf1bd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-11 14:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db965be04295eebd8192444ed3df12c0a46a9ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 14:34:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a417b4f6ea439ea5ddca5138e5358b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/836d27544f434c5eaca3b55d2aec517f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae45c6730184a699da3c406f760215c : SUCCESS in 7m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b477a5a4512c4aaa90911ae613196c8a : SUCCESS in 7m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/81793e224ce74066b9ce206f393535bb : SUCCESS in 3m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ceb2d30cab8f41aea0d6446317c52718 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5782c53b2b7231c27ca623ee76f7e8ce1080c950","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-14 14:00:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ff7e668ec33bba7699dcaf7508905f62d825c1b6","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-18 23:07:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f56922f9d06ff1822887e061830b8493bd8d2fb","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-18 23:07:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d77fb1fcb04f20e9bc1005626cc97b29cb7b03ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:08:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fcf048d14a3f15115a31251c507973b0ea2afd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:26:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/983aeb535b824eb09adebe41934115e3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d0d67f736945549d977f96d3bea8d1 : RETRY_LIMIT\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72b379c685d443fbbfe7ad0775f629a : RETRY_LIMIT\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b30fb53ce6c483d98c49424fb0773e1 : SUCCESS in 5m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/aa2647e5245d4bb58ec187f03b3d6313 : RETRY_LIMIT\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/22ae79811f8041baac04c578d6cdf865 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"617dc77041c8aa85ff49531a011bf46af3253c36","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 23:26:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1b26eb94bbd9a795d3704a1854309547e78ed30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bf10dd138d74f2eb3064dffdf125a44\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e52cd14a696a4bce9b98d7d1ac90412c : RETRY_LIMIT\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7241b082b26941a5973e65cb61b01bbc : RETRY_LIMIT\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1efdcceaef5d414eaeb57a9f301e232f : SUCCESS in 11m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0a439a83e5b84339a43f375b4cde8a6b : SUCCESS in 3m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1b712c1e430d46de879ec1478faa429b : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ed2964c41dc8fc4782c161cba8e26abb5cda7d31","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-21 09:44:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b869e51f72f5b6ee191cbc6a3cec193ccfd6612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84cb77b9e00d4c64bc3a1aadd13fad7f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c70f77170e0407dbda4946b6810a975 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c76a25d7a3047878f2a12de9a53140a : SUCCESS in 7m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/93d04a1d8b4d4eecabe8a7189748d15d : SUCCESS in 12m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c8ef6a72dff24f9986fb81fc62d8ecf0 : SUCCESS in 6m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a5a0f095b2374006a9688c9fc37731de : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/06c48fc8d6dd4039b5988b7519a06290 : SUCCESS in 1h 13m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5b0415748ea14f37a7a38dfffb380b58 : SUCCESS in 1h 03m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4c0eea3301654a9abf53041cce3238b7 : SUCCESS in 56m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5f0df14104d14898b647bf975b73c5c9 : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ba49ea82b20e01441ca0f2eb87f3e07db4c4660e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:22:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78eac3892c1470314b0ea603f22101596e6e6846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:36:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e645798771b4042bcb3b0160bde0c3b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ebd84c2246458ca27b4e05dae53f73 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f6ef0435384a64b36570c05a4afaa8 : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3a3c61d35c5b434889588446a8ac9bc3 : SUCCESS in 11m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7371920e937b45c08291f1dbd25249e6 : SUCCESS in 7m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/220999611e634dc0afd3dd8c33776a05 : SUCCESS in 5m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6d5f5d6613a544689edbb5f88fc7a2b1 : SUCCESS in 1h 08m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/29548924fe124ec58e754af9ad7c0a2d : SUCCESS in 33m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7f644d2da02b424f8197c839992694af : SUCCESS in 53m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e83c6fc1d57747298fc764baa1bce8c7 : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4787a3d50be6a6cbbbd0976c9506a640066ed655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:36:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da9bf1f48e9796bf6e99c05595ddf5ab6a369256","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:37:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f9d52b389f40f39be0655bee6cc93c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c9e33cd4ed4453ac25c704f267cb51 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6245b6cb6d8e405bd93bd9dc7721ce47fb64194b","revisions":{"6245b6cb6d8e405bd93bd9dc7721ce47fb64194b":{"kind":"REWORK","_number":1,"created":"2026-09-11 14:16:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/59/1005259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/1005259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/1005259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/1005259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/1005259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/1005259/1"}}},"commit":{"parents":[{"commit":"afbd8ab93f113cd7a40fe1e3ad6e3174baa0cd41","subject":"Merge \"Add German (de) locale\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afbd8ab93f113cd7a40fe1e3ad6e3174baa0cd41"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-11 14:16:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-11 14:16:22.000000000","tz":0},"subject":"Update .gitreview for stable/2026.2","message":"Update .gitreview for stable/2026.2\n\nChange-Id: I7949a54e1ae6a3260ceb3ca0b383cb6700c1bfe9\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6245b6cb6d8e405bd93bd9dc7721ce47fb64194b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6245b6cb6d8e405bd93bd9dc7721ce47fb64194b"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
