)]}'
{"id":"openstack%2Fskyline-console~1005261","triplet_id":"openstack%2Fskyline-console~master~I85e98628f842df8ceea426a5200a24e5b09c1e83","project":"openstack/skyline-console","branch":"master","topic":"reno-2026.2","attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-09-14 14:00:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-09-18 23:05:30.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}},"hashtags":[],"change_id":"I85e98628f842df8ceea426a5200a24e5b09c1e83","subject":"Update master for stable/2026.2","status":"NEW","created":"2026-09-11 14:16:32.000000000","updated":"2026-09-18 23:32:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d90b166e07f5f0f6456d9f2bd2393c953ba0d25","_number":1005261,"virtual_id_number":1005261,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 23:32:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2026-09-18 23:05:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-14 14:00:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2026-09-18 23:05:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 14:37:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-14 14:00:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-09-18 16:35:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"b4a74dcae7b3585508c4770a301f89aeed87f8e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-11 14:16:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0dbb32a458357f251452575803a05d065449753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 14:37:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d1ca6217c748c1a50dc1f7f4f10f27\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528f5ca26d1043949271759d1eac9c95 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7378f18a37c9450ca40e77a8a0925e1d : SUCCESS in 7m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9aa9c336f1b64555af62d6e2a28d72b7 : SUCCESS in 11m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0b829b99a6964134973405bfd9a5e734 : SUCCESS in 6m 11s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f07dc8f393294ffdad03a566e6c67a42 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"69a7adeacac988bbbac7c61c3aa1c28869e9262f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-14 14:00:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9e85b914fbaf9aa284e760e50540d8d6d950fcf","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-09-18 23:05:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0499cc6c1404994b101a7ca0d719d650ba91fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:05:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fb88dc0d898edbd3d20610e96b14fdbe0fd2815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:19:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5aaba096e89c4efb980c6fddb52d91c7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484bb64bfb4c46ca93603f1ada3bfaa2 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d7bded932b46a6bd08b3d816016b68 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74b3aaf5a01b43328905404700cdcf1b : RETRY_LIMIT\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e7582c1389dd4df798ca0084ef481e2d : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9c011ed771ad458e97fb3ecd1427e252 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"90142c81aa174107f6439bcd2402793fe389720c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 23:24:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d90b166e07f5f0f6456d9f2bd2393c953ba0d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 23:32:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28e056a7be244d0b97984823a3ba672d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e10059972e243a5b124c8c3e14bf516 : RETRY_LIMIT\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd8e7ae91404e7ebafffa833719db9d : RETRY_LIMIT\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/126a05b96e9d46fbb3fff9acd07dd1b7 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5822b55d6bfe420ebaecf3e1cfa0f7e1 : SUCCESS in 7m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/33e88e5c714949bcbcc97e3587b98b75 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ee0376f6b53f47eec5afc036f049b04b2bb0364","revisions":{"6ee0376f6b53f47eec5afc036f049b04b2bb0364":{"kind":"REWORK","_number":1,"created":"2026-09-11 14:16:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/61/1005261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/61/1005261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/61/1005261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/61/1005261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/61/1005261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/61/1005261/1"}}},"commit":{"parents":[{"commit":"e3f7d1dba56f0f8411ce12ef11639c5050f114a1","subject":"Merge \"Make Magnum cluster nodes clickable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3f7d1dba56f0f8411ce12ef11639c5050f114a1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-11 14:16:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-11 14:16:28.000000000","tz":0},"subject":"Update master for stable/2026.2","message":"Update master for stable/2026.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2026.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2026.2.\n\nSem-Ver: feature\nChange-Id: I85e98628f842df8ceea426a5200a24e5b09c1e83\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/add_release_note_page.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6ee0376f6b53f47eec5afc036f049b04b2bb0364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6ee0376f6b53f47eec5afc036f049b04b2bb0364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
