)]}'
{"id":"openstack%2Fskyline-console~851703","triplet_id":"openstack%2Fskyline-console~master~I5b3972dc3de3b579aed5f3ce798cfe4449c30d1a","project":"openstack/skyline-console","branch":"master","topic":"quota","attention_set":{},"removed_from_attention_set":{"26304":{"account":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"last_update":"2022-08-03 01:59:14.000000000","reason":"removed on reply"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-08-03 06:33:34.000000000","reason":"Change was abandoned"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-08-03 06:33:34.000000000","reason":"Change was abandoned"},"30356":{"account":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"last_update":"2022-08-02 07:17:52.000000000","reason":"removed on reply"},"34458":{"account":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"last_update":"2022-08-02 07:18:24.000000000","reason":"removed on reply"},"35102":{"account":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"last_update":"2022-08-02 07:17:45.000000000","reason":"removed on reply"},"35103":{"account":{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"},"last_update":"2022-08-02 07:17:15.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5b3972dc3de3b579aed5f3ce798cfe4449c30d1a","subject":"fix: fix the size quota check when create volume","status":"ABANDONED","created":"2022-08-01 08:45:04.000000000","updated":"2022-08-03 06:33:34.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7310e92bfbe72a95996f994d9e1dc2e98cc1a40d","_number":851703,"virtual_id_number":851703,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-02 08:56:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"},"all":[{"value":1,"date":"2022-08-03 01:59:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":1,"date":"2022-08-02 07:17:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2022-08-02 07:18:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-02 07:17:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":1,"date":"2022-08-02 07:17:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"}],"CC":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-01 10:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 02:58:11.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"},{"updated":"2022-08-02 07:17:13.000000000","updated_by":{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"},"reviewer":{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"},"state":"REVIEWER"},{"updated":"2022-08-02 07:17:44.000000000","updated_by":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"reviewer":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"state":"REVIEWER"},{"updated":"2022-08-02 07:17:51.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2022-08-02 07:18:23.000000000","updated_by":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"reviewer":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"state":"REVIEWER"},{"updated":"2022-08-03 01:59:13.000000000","updated_by":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"reviewer":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"state":"REVIEWER"}],"messages":[{"id":"f81ed42641a8067a74a8ca55de24f4c5ff7a1fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-01 08:45:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5776d3e22a2504075e6085c4bb19a5f41c635900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 10:23:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9d3f3d1557d44b6ab279c43131593ac\n\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1286805561de48ac933f6c575aa3a8d1 : SUCCESS in 8m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a102ffe3838b40d69a975c6e9bf05074 : SUCCESS in 9m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0bf33bc201874ed2999db4ae865b1526 : SUCCESS in 8m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a974cb9a9d354f53be07bd2c9a0fbe11 : SUCCESS in 1h 35m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/205d99a8614a45de8bdf7e7da39df8ea : SUCCESS in 1h 10m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3bda2bc5cc764a359703c7e60fd7d2bd : SUCCESS in 1h 01m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4592c019f8a64385aafc5353578e4436 : FAILURE in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d784ba267cda989885acdd80946b590a12b381c4","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-01 10:50:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dec3d23c4d6ef573fe3548c5d1b4d9fcc52c9e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 12:32:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7ecdae331bc443bb7f96ec769ce75d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3568bb3fa54441bf1bab6824dfab8c : SUCCESS in 7m 45s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d7286d3342bb4852b2c41f8059972fec : SUCCESS in 10m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/54096357909746cab74508e0bb2f49a3 : SUCCESS in 8m 40s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4e6a9ba09ded473abcd9dcefe66ffba2 : SUCCESS in 8m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/84bceeb0b0de4a76b7510141460519be : SUCCESS in 1h 34m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ba275446a2e4468fac9221eb8b38c3eb : SUCCESS in 1h 10m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/26c4748257e44e149616ddf751f2889d : SUCCESS in 1h 02m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/65f1de8436cb47c19f3983e397c1a9fb : FAILURE in 1h 04m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cbd9b6037b9b3b80db48b65ab7a442b40f6b272f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-08-02 02:58:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08e29c267b5e1d7082e3c4ee098c64fdf326cc87","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-02 02:59:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb7d16d1b525723103389b7fa26692a365a603c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 04:38:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bb32092dfe24546a0eeb3c2703db831\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe39ab62dc247ebb6a3d5edceb61676 : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5787d6cb0b4b434fb460a62e84c4514f : SUCCESS in 8m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/996c9a27a37a4ec098e602ccbe50b96c : SUCCESS in 8m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5486a2cf922541149fe76437a50404b7 : SUCCESS in 8m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ea5966e0fc514a829b0f5a62fbf88b29 : SUCCESS in 1h 33m 38s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/230fa15807de4837a9623c0b1d57b3c6 : SUCCESS in 58m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/990966ce97904489ba35ab704255f692 : SUCCESS in 52m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a5066e677b0346a6be1988c62686bfba : FAILURE in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7683377ccc61f01d263f93ed9010191ea9ea4d3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-08-02 05:08:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5ea87875e850df02fdad2c6c4b5aade04e32cad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 06:52:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63c7bc5f44f349bf836d7fb7bf00001c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a490a227f9044bcb879bfaee56bc1164 : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fdabd54ef27d4b7ca25a178e1acc5018 : SUCCESS in 11m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d0e88f3009c40bd895db95c585993f2 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/87b581ab2e6f4bea99e7914560b65c28 : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9d6952aecd0345d798faa6e362e9e799 : SUCCESS in 1h 36m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f3cefbaed5ff4b89be6f3bba0e7df94f : SUCCESS in 1h 10m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0785cb4dbaac46f29b425535b18bd905 : SUCCESS in 55m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9b682beba5a14d638dfae48d1905e325 : FAILURE in 52m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5b86063591255ea326431292945e0cd3384fa84d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-02 07:14:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1f8baa1316e1fcb39d51760cd81d7b79e0fd041","author":{"_account_id":35103,"name":"wanya","email":"wan.ya@99cloud.net","username":"wanya"},"date":"2022-08-02 07:17:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b387dc17e44718bf84e937a15cf99587a7ad1c2b","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-08-02 07:17:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"624e2aa29a84e49f37a380792c7166c8da042ccb","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-08-02 07:17:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"faa40bca23605d1a41f66ffc7a03886ae81e645d","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-08-02 07:18:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2a6c2fc1b36cb0d21a52d07026ca54dcb4abf9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 08:56:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab09938bad884e63aa8c4ad0bfdfc2ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf31b1c9afb40e6861e7c7667fb2361 : SUCCESS in 5m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/446a5f5773ae48d2b80fb4138eb85cfc : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1f74ff6b9cdb40e78a2f265bb96a4721 : SUCCESS in 8m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4c3cf7b301a844f5bbaddb1e9e6babc1 : SUCCESS in 8m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/194364ee1a8d4e55a5140b252f6d34c9 : SUCCESS in 1h 37m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/30d726d1ecb94e158f291c14a0c8135d : SUCCESS in 1h 08m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b1b70c1d723f4de1be6045b5f8e77346 : SUCCESS in 46m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1f0ed2fbfce0422ebd4a6acf4fd90cee : FAILURE in 1h 13m 23s","accounts_in_message":[],"_revision_number":3},{"id":"59af569ebb4616c731f884648dfde17e559318bf","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-08-03 01:59:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7310e92bfbe72a95996f994d9e1dc2e98cc1a40d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-03 06:33:34.000000000","message":"Abandoned\n\nabandon","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"404cbeb95fb2e40ab33760f6437cb140118c7b3e","revisions":{"be945585a9715590985542eb4ab7c3be50afff2c":{"kind":"REWORK","_number":1,"created":"2022-08-01 08:45:04.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/03/851703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/851703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/1"}}},"commit":{"parents":[{"commit":"c121385eda9e124ccb46f836bdc9dc746abf4d3b","subject":"fix: fix instance quota check when create instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c121385eda9e124ccb46f836bdc9dc746abf4d3b"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-08-01 08:43:59.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-08-01 08:43:59.000000000","tz":480},"subject":"fix: fix the size quota check when create volume","message":"fix: fix the size quota check when create volume\n\n1. Disable to click submit button when the remaining size/type size quota is 0.\n2. Update SliderInput component to onChange the right value, not the incorrect input value\n\nChange-Id: I5b3972dc3de3b579aed5f3ce798cfe4449c30d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/be945585a9715590985542eb4ab7c3be50afff2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/be945585a9715590985542eb4ab7c3be50afff2c"}]},"branch":"refs/heads/master"},"d62118a6f56c8895bc71873932a9911a2c31c370":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-02 05:08:59.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/03/851703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/851703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/2"}}},"commit":{"parents":[{"commit":"37885d343dcdba31e6ecc3c6c33e2659a20dac41","subject":"Merge \"fix: fix the plural form of the menu words \u0026\u0026 tab title words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/37885d343dcdba31e6ecc3c6c33e2659a20dac41"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-08-01 08:43:59.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-08-02 05:08:59.000000000","tz":0},"subject":"fix: fix the size quota check when create volume","message":"fix: fix the size quota check when create volume\n\n1. Disable to click submit button when the remaining size/type size quota is 0.\n2. Update SliderInput component to onChange the right value, not the incorrect input value\n\nChange-Id: I5b3972dc3de3b579aed5f3ce798cfe4449c30d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d62118a6f56c8895bc71873932a9911a2c31c370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d62118a6f56c8895bc71873932a9911a2c31c370"}]},"branch":"refs/heads/master","description":"Rebase"},"404cbeb95fb2e40ab33760f6437cb140118c7b3e":{"kind":"REWORK","_number":3,"created":"2022-08-02 07:14:17.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/03/851703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/851703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/851703/3"}}},"commit":{"parents":[{"commit":"37885d343dcdba31e6ecc3c6c33e2659a20dac41","subject":"Merge \"fix: fix the plural form of the menu words \u0026\u0026 tab title words\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/37885d343dcdba31e6ecc3c6c33e2659a20dac41"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-08-01 08:43:59.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-08-02 07:10:27.000000000","tz":480},"subject":"fix: fix the size quota check when create volume","message":"fix: fix the size quota check when create volume\n\n1. Disable to click submit button when the remaining size/type size quota is 0.\n2. Update SliderInput component to onChange the right value, not the incorrect input value\n3. Add image check to the volume e2e after created image successfully\n\nChange-Id: I5b3972dc3de3b579aed5f3ce798cfe4449c30d1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/404cbeb95fb2e40ab33760f6437cb140118c7b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/404cbeb95fb2e40ab33760f6437cb140118c7b3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
