)]}'
{"id":"openstack%2Fskyline-console~858614","triplet_id":"openstack%2Fskyline-console~stable%2Fzed~Ie84ef2aac98da92c5707db9e0355e23f3e96f6f9","project":"openstack/skyline-console","branch":"stable/zed","topic":"bug/1990213-stable/zed","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-09-21 06:41:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie84ef2aac98da92c5707db9e0355e23f3e96f6f9","subject":"fix: set AZ required when create aggregate","status":"MERGED","created":"2022-09-21 06:15:19.000000000","updated":"2022-09-22 03:31:18.000000000","submitted":"2022-09-21 06:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858614-bug/1990213-stable/zed","cherry_pick_of_change":858485,"cherry_pick_of_patch_set":2,"meta_rev_id":"1775303d543e6e72d7f1c102f6ba883269899516","_number":858614,"virtual_id_number":858614,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 06:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-09-21 06:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-09-21 06:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-22 03:31:18.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-09-21 06:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 06:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 06:29:50.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-09-21 06:38:38.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-09-22 03:31:18.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"109ee69b85a800ae7bcc6fe0834e326be570acf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-09-21 06:15:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e3588e1eef6a270b99813444ea760a4b3beef5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f6ec01d8d5f48398c233704815b1785\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5a844e2c6b44ec84f765cb9b355a7b : SUCCESS in 7m 47s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/440e77fd08d14b8a8feff8cf83197f06 : SUCCESS in 11m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dde2283217e14d24a618fbdc9b036ba0 : SUCCESS in 9m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f74b559acf4f4c5384871afb5a49e11f : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f169178d54cc839c263467000934b0a6800d0b01","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-09-21 06:29:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a69878c7d9c9aa4cae04b97c6eca753f29667978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfa056d3a1d79e4ffd40209de04a756fb05d7a40","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-09-21 06:38:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bedb637d8dc0b8626f8959e944e2e2e6c0c369b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:41:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde20515cc3f407a83be93be0586b092\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b7915858d1412a8e658b2ca8c824bb : SUCCESS in 7m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b9263294e289475d86be40a3da854016 : SUCCESS in 10m 19s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5cda4e43a2724cb299b64412d5ade949 : SUCCESS in 8m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b3da9c9eec7d427aa269d4e0725d8f30 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a05e98b4e43eeb7224365a598121ffae2df743e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:41:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4188a73c6ec0a975614d899cc34dd29cc7ce5764","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:42:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de1b5049bf214c889b5499331fd4d36b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526b8332278f4b66980f64ec1d8173e5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"1775303d543e6e72d7f1c102f6ba883269899516","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-09-22 03:31:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bec03e7fdffd3c6c6bd344c64a807654ce8a1e10","revisions":{"bec03e7fdffd3c6c6bd344c64a807654ce8a1e10":{"kind":"REWORK","_number":1,"created":"2022-09-21 06:15:19.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/14/858614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/858614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/858614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/858614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/858614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/858614/1"}}},"commit":{"parents":[{"commit":"3d7324765a02b6c29445ab81d35e65bcf54fefa6","subject":"Merge \"fix: Fix the unit when upload file\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3d7324765a02b6c29445ab81d35e65bcf54fefa6"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-09-20 05:32:18.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-09-21 06:15:19.000000000","tz":0},"subject":"fix: set AZ required when create aggregate","message":"fix: set AZ required when create aggregate\n\n1. Set AZ input required when create aggregate with new AZ\n2. Set AZ select(which contains a no-select AZ) required when create aggregate without new AZ\n\nCloses-Bug: #1990213\nChange-Id: Ie84ef2aac98da92c5707db9e0355e23f3e96f6f9\n(cherry picked from commit 849db0f740b3c965422c2b1a4797f83c3b7b9ff6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bec03e7fdffd3c6c6bd344c64a807654ce8a1e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bec03e7fdffd3c6c6bd344c64a807654ce8a1e10"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
