)]}'
{"id":"openstack%2Fskyline-console~858615","triplet_id":"openstack%2Fskyline-console~stable%2Fzed~Ie0e6e6171d69bfa63f49727ffdc0626b6266eb80","project":"openstack/skyline-console","branch":"stable/zed","topic":"bug/1990241-stable/zed","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-09-21 06:41:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0e6e6171d69bfa63f49727ffdc0626b6266eb80","subject":"fix: fix fetch subnets","status":"MERGED","created":"2022-09-21 06:15:47.000000000","updated":"2022-09-22 03:31:12.000000000","submitted":"2022-09-21 06:41:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858615-bug/1990241-stable/zed","cherry_pick_of_change":858598,"cherry_pick_of_patch_set":1,"meta_rev_id":"5877ee72043ff805d18953265bccf61508195148","_number":858615,"virtual_id_number":858615,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 06:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-09-21 06:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-09-21 06:38:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-22 03:31:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-09-21 06:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 06:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 06:30:00.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-09-21 06:38:32.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-09-22 03:31:12.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"98d3d2f251029c080c965dbde57ed371d8930625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-09-21 06:15:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"abbdf42b9aca72bc2b01b9168e01953acfb516b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8f936292af43a89e0405a04a4cf2cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c61067602574e4eac3265e6eb7660cd : SUCCESS in 5m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cd8b3ad45d0d4492aadf128aa2663b14 : SUCCESS in 7m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ee8c337c5ddb4d52977910d58e9f6b0d : SUCCESS in 7m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/181db842dc994fd0a8005bfa8538bc73 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"de4f940602c4163fa69c7a245af9722dd0bb1e68","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-09-21 06:30:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f43ef85473872956ab979c51ee9becbb7100059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:30:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"266f749730d3521c181ce48d19c806147cc341d6","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-09-21 06:38:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e23eb90fa3a869e2c3d70a318a85c349398f6c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:41:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aebfcc12734a4b49b4513a66069b3f74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a0af99f19934c4baedf5af8d7838d8d : SUCCESS in 6m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6e2c614c12ac4a9ca179ae5335621aa9 : SUCCESS in 8m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8ee226f0e65b4d499217100951f07eb2 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/44f022ade80440fb9aa00f1a3701cf34 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"845a0d7a45b5d81dcd2c05438d11e27c65c6d19f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:41:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b5757cb984aa18be2f1f4844b7f032ebcdcf527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:43:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed94612b488c4437b70e35336838e23e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7ac327cc24498983a20b0c95a2b015 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"5877ee72043ff805d18953265bccf61508195148","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-09-22 03:31:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f12cb36aaea5da5d4643b42ce8b2b0de0908e4ca","revisions":{"f12cb36aaea5da5d4643b42ce8b2b0de0908e4ca":{"kind":"REWORK","_number":1,"created":"2022-09-21 06:15:47.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/15/858615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/858615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/858615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/858615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/858615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/858615/1"}}},"commit":{"parents":[{"commit":"3d7324765a02b6c29445ab81d35e65bcf54fefa6","subject":"Merge \"fix: Fix the unit when upload file\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3d7324765a02b6c29445ab81d35e65bcf54fefa6"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-09-20 08:45:44.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-09-21 06:15:47.000000000","tz":0},"subject":"fix: fix fetch subnets","message":"fix: fix fetch subnets\n\nIn adminstrator, remove the request parameter all_projects when fetch the subnets, because the subnet list api in the Y/Z not support this parameter\n\nCloses-Bug: #1990241\nChange-Id: Ie0e6e6171d69bfa63f49727ffdc0626b6266eb80\n(cherry picked from commit 06af9d6fd379ef6855a773cde44d634129935329)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f12cb36aaea5da5d4643b42ce8b2b0de0908e4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f12cb36aaea5da5d4643b42ce8b2b0de0908e4ca"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
