)]}'
{"id":"openstack%2Fskyline-console~861143","triplet_id":"openstack%2Fskyline-console~master~I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b","project":"openstack/skyline-console","branch":"master","topic":"bug-1992739","attention_set":{},"removed_from_attention_set":{"26304":{"account":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"last_update":"2022-10-14 01:41:06.000000000","reason":"removed on reply"},"30356":{"account":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"last_update":"2022-10-14 01:40:30.000000000","reason":"removed on reply"},"34458":{"account":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"last_update":"2022-10-14 01:41:34.000000000","reason":"removed on reply"},"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-10-19 11:22:30.000000000","reason":"Change was submitted"},"35102":{"account":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"last_update":"2022-10-14 01:40:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b","subject":"fix: Not show the data disk when creating vm by Instance snapshot","status":"MERGED","created":"2022-10-13 07:15:51.000000000","updated":"2022-10-26 01:30:47.000000000","submitted":"2022-10-19 11:22:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861143-bug-1992739","meta_rev_id":"e1ea1a1fbdfda70e3157791d899c1b193519f6ad","_number":861143,"virtual_id_number":861143,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 11:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-10-19 09:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":1,"date":"2022-10-19 09:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":1,"date":"2022-10-26 01:30:47.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-10-19 09:06:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 08:36:26.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-10-13 08:40:26.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-10-13 08:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 01:40:28.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2022-10-14 01:40:57.000000000","updated_by":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"reviewer":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"state":"REVIEWER"},{"updated":"2022-10-14 01:41:05.000000000","updated_by":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"reviewer":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"state":"REVIEWER"},{"updated":"2022-10-14 01:41:33.000000000","updated_by":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"reviewer":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"state":"REVIEWER"},{"updated":"2022-10-14 02:56:00.000000000","updated_by":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"reviewer":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"state":"REVIEWER"},{"updated":"2022-10-14 06:29:59.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2022-10-14 06:30:30.000000000","updated_by":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"reviewer":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"state":"REVIEWER"},{"updated":"2022-10-18 01:45:58.000000000","updated_by":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"reviewer":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"state":"REVIEWER"},{"updated":"2022-10-18 08:08:00.000000000","updated_by":{"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},"reviewer":{"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},"state":"REVIEWER"},{"updated":"2022-10-18 10:20:26.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-10-18 10:27:51.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-10-19 01:47:11.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-10-26 01:30:47.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"4d858f42cfa719ec837422c1cca2acacb290429b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 07:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2f27004d290d516da011a992a3a9eaf112d0a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 07:16:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4174a88c0911d88eb0adc1a3afe2d33a59e3a6cf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 07:17:57.000000000","message":"Topic set to master(bug/1992739)","accounts_in_message":[],"_revision_number":2},{"id":"ebe98383b08309ded59df5690e1b888476ed7537","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 07:19:28.000000000","message":"Topic master(bug/1992739) removed","accounts_in_message":[],"_revision_number":2},{"id":"b7434b3be28196e0eea87e2e19eaa0727156e5d0","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-13 08:36:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"be144ea9aac5a7e7176ab432c7b7e049b6302172","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-13 08:40:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2cb752f291e0ba5a991536f8a6bbfcecbb4f8391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 08:43:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6f011037c144a5ca9ddef12d83b4a7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b32101c1c5438e81dd8093f3d9f09a : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f02933e3f98343faac0c8bf12cff3530 : SUCCESS in 6m 47s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f65330d7f4494a6c927c7c88df7257d5 : SUCCESS in 6m 33s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9c9a1b9e9ef74099a4749521cf969398 : SUCCESS in 4m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d395604b980c4bb9ad52ec92bb26ed48 : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4819215fefe3b36d83fc347b73947e25fe0146d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 08:50:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"264b2cbb45c4be31ede3dff2c8835448a3caeaf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 08:50:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"68803087e33736a518e58ed5c8da528ac6e6bf6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 09:21:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1a235c1aa649213b8c5f21d9788350fc98ccd014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-13 09:36:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bea87bd84628000bb0f9117cd3186c1f0ddb1b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:17:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2805b9855b4095abe2b68fb7890178\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f83df734a04734bdc58d32c2816217 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/12714e44617244eb99bcb11bcdcec0db : SUCCESS in 10m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/72b15d8e327a40d895e57ec3618aae92 : SUCCESS in 6m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f522a08e78f44c348fc9d008f33257c6 : SUCCESS in 5m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/64a6e4f05f9147de84eca30602e8af98 : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":6},{"id":"64dd4a21a702ea14fa85b22670eb7874cf60b9fd","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-10-14 01:40:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"13c2c780bd7713005385b3aea231010a76cd32d9","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-10-14 01:40:58.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"91c2a36c121642e63db543d3aef0c76a2ba28591","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-10-14 01:41:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8359f24c39cc83306d1c551211826de0eb01e1f0","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-10-14 01:41:34.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"921530fc74d16ce5996d589027a1d44708c78e51","author":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"date":"2022-10-14 02:56:00.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d1544fad8946d0e5bf649728af4b0cc2fca67e90","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-14 02:57:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"419184f63e81e3f4c86498e4697f9ce3d1f734cc","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-14 05:54:56.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7935f97ec5f3955686f0fb0e6b5e76d6fe724b41","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-10-14 06:29:59.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c19076f1821cd35356f0e5fec7f071d4635fde43","author":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"date":"2022-10-14 06:30:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"133f6104ad7680337d6d3fa081f11d213b5067ca","author":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"date":"2022-10-18 01:45:58.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"357d5eb750f0fb738dfc498ee71c31c2ed21700b","author":{"_account_id":30576,"name":"hui.chen","email":"woraser@gmail.com"},"date":"2022-10-18 08:08:00.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e9b76b02dff9e347c48ae41b693464d5ba9082ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-18 09:07:05.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"10a0e6dc029b3618b0fcf94476bf0293d8469ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-18 09:28:36.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"dd6a363aaed632705054a163d10ce6cd73f8c22b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-10-18 10:20:26.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"259ebdcc39d8fca4604439c81ae22e923503e346","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-18 10:27:51.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1f8c864153de474e3c9ceafc3cf87a79dfc88db7","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-18 10:29:33.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_28706\u003e\n","accounts_in_message":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"_revision_number":8},{"id":"8c0ab765f8110f60e8038a82038d132f85deb363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:52:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd8115742a8c400088dcdfa1592145cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f52577480b412a8eb173d828324c38 : SUCCESS in 5m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/be653923f0754d71be14d749b1dced79 : SUCCESS in 7m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7cdd248ca13341329dac75bc92207ea5 : SUCCESS in 6m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/04ec49811b3c4246afc4aba3eeb23256 : SUCCESS in 7m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/aaf78cca3e49439fb197859585928807 : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":8},{"id":"ae6aad2451013333d15cdd0aa2a3a915f8826350","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-10-19 01:23:22.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e2076e4e08b71c0078a6f8d331fffe789b118313","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-10-19 01:24:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"3701d51c5d0706d038851092d29828b68c4a8b7f","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-10-19 01:24:42.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"68667cec15fe422cb6a78248f08646b5d6b16a5e","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-10-19 01:25:31.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"15c8f63c7509b7643c077eb9e280e08a57276bd5","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-10-19 01:47:11.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"bcfd26277010823be9c7a7c04272e78c25115a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 03:16:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5750780251759f58e76c683713896780ec7a5392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 04:03:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"804ec03c1e45b25e2b584c606cd0824ff843c472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 04:12:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7fb8cbb37090d5f92d3eb652f72a95d9c84bf15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 05:23:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0598bf2a57524809859e526e25c90f8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b449a59927ee4c1a8db6a8638326afc7 : SUCCESS in 4m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dae3c99137344680bc0638119f09ed8c : SUCCESS in 6m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7cb824c4d06649f080183759872e825b : SUCCESS in 5m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/17856a4c4a9241d68d2aa3160773fbd6 : SUCCESS in 5m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fdef5bf3e5f84316a1518d0325ad0390 : FAILURE in 1h 07m 41s","accounts_in_message":[],"_revision_number":11},{"id":"d946fb1232f556f9e571a6f53fa8903d6e92bf2b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-10-19 06:00:50.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"26ccf2ad2f01f8cd6a1ee4515d9d89efd5dfbcd4","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-10-19 06:01:49.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"9fcaf565d57353a6ae933d094841fde136c1e935","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-10-19 06:02:06.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"2c2b9b0e8baef97b56aa9df754a71302ae559b01","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-10-19 06:03:06.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"a372f3f7712f829afe46feff56a31a484f8bdcaa","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 06:22:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"934ae0dd449083789bbf2926b688c6a62b89b41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 07:49:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b18c1b497ac41f981a98b295a6b1ffb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94937025e21462996666a99069b36bc : SUCCESS in 4m 08s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/da448c2e383248cc8a64a272515eebda : SUCCESS in 7m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b12a20e854e541d8a676733e4fdbe00f : SUCCESS in 4m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/928f40bdcc5c4cec9f2d3293fc42892f : SUCCESS in 6m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ee2bd006bf7a43b0b26b58f8b9dbdb4a : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":11},{"id":"b95adaca25e601960f51c894d944e684cebd2f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:07:27.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f0303adfaa0caec6c1e3722a5e133f65f3b9daf4","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-19 08:10:49.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"fa9306de676ae481f682411c2bfaf32aa6a7f66c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-10-19 08:27:52.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"21eaf85a1d8136579e984db616fca2d0d636cb11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:28:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c89d206bffbd5c764e4d17cf574ca97fd4014756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:40:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"fda8335ea0fefd0372203bc38f168d1290848271","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:43:48.000000000","message":"Topic set to bug-1992739","accounts_in_message":[],"_revision_number":14},{"id":"f9c36376f4103469571506c99e4387d923ab6e4a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-19 09:06:51.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"0f16468e78aecb870e6e2742900746c298275d91","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-19 09:11:42.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"2129e8cbc6a14f60b2bed6bdb7edc65f14803835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:02:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba44bd7a188c437485289d2b85d6e17d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64186cf59f8f4426aa2c1d60487f6272 : SUCCESS in 4m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ec232bb9af3044c7bae8da3546114579 : SUCCESS in 9m 34s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d93aa593d8304045b47fe31cb7c6706a : SUCCESS in 4m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b3eb37a90d3946398265b380e804d464 : SUCCESS in 6m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/674e11a2a344495eb8a8879e01bb2ca4 : SUCCESS in 1h 19m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dbb34b001b1c4ddfa5d65ae8354991e2 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":14},{"id":"86372c0db1ef61084eba2814be388e48cb4bffd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:02:52.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"b67bbe90e9ffe7fb2d474227882667f907472394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:22:30.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/141267fb21334ecda015d99cb76c6e5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3945c7800684b12aac837a45a4cffb9 : SUCCESS in 5m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a99e49baaff64f219941c56e067036a3 : SUCCESS in 8m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7ce0178b679248958f1e851ace58433c : SUCCESS in 6m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/694b64d55b7049bdb6aa8a1c81d7eefd : SUCCESS in 6m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8bcc846862f749db8995ea70bd71066a : SUCCESS in 1h 18m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/86f5ac818d874e10bb3809780fce7db1 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":14},{"id":"a1778fa23a09b7f9705844b18bf934f12c35998b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:22:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"e249d13704a15f7d2cc0046d38d66765397b6c26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:23:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae482bba49c4cf7b2bd161b864b46e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425546d0f67b4b3ea164d530c1d5bb95 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":14},{"id":"e1ea1a1fbdfda70e3157791d899c1b193519f6ad","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-10-26 01:30:47.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"fdd6dbb5e38e63bad84f9dc68a13e2a18b3147be","revisions":{"e017b3540025f0e0cbe1977c171facdc7848c29f":{"kind":"REWORK","_number":1,"created":"2022-10-13 07:15:51.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/1"}}},"commit":{"parents":[{"commit":"86937127c9b610f128c82d8780cc2182c0952339","subject":"Merge \"fix: set the default language is Engilsh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/86937127c9b610f128c82d8780cc2182c0952339"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n\n1. Go to Compute-Instances page\n2. Click Create Instance button，create the instance with 1 system disk + 1 data disk\n3. Create Instance Snapshot by the instance by step2\n4. Create Instance by the instance snapshot by step3\n  4.1 Select Instance Snapshot of Start Source\u0027s option\n  4.2 Select the instance snapshot by step3\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e017b3540025f0e0cbe1977c171facdc7848c29f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e017b3540025f0e0cbe1977c171facdc7848c29f"}]},"branch":"refs/heads/master"},"52d69cbec1ec7faf95741a1053754b972b6aaea9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-13 07:16:40.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/2"}}},"commit":{"parents":[{"commit":"b24698f3e23499303f171777441c81f9fc10222a","subject":"Merge \"feat: Support for custom prometheus requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b24698f3e23499303f171777441c81f9fc10222a"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:16:28.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n\n1. Go to Compute-Instances page\n2. Click Create Instance button，create the instance with 1 system disk + 1 data disk\n3. Create Instance Snapshot by the instance by step2\n4. Create Instance by the instance snapshot by step3\n  4.1 Select Instance Snapshot of Start Source\u0027s option\n  4.2 Select the instance snapshot by step3\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/52d69cbec1ec7faf95741a1053754b972b6aaea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/52d69cbec1ec7faf95741a1053754b972b6aaea9"}]},"branch":"refs/heads/master"},"07879e53f23a00014c9edabaf354c47f8f3901a3":{"kind":"REWORK","_number":3,"created":"2022-10-13 08:50:03.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/3"}}},"commit":{"parents":[{"commit":"b24698f3e23499303f171777441c81f9fc10222a","subject":"Merge \"feat: Support for custom prometheus requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b24698f3e23499303f171777441c81f9fc10222a"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 08:41:33.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n2. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/07879e53f23a00014c9edabaf354c47f8f3901a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/07879e53f23a00014c9edabaf354c47f8f3901a3"}]},"branch":"refs/heads/master"},"3fff81dc66982e8f2093ef2a08a5d7a9cabe83d4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-13 08:50:35.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/4"}}},"commit":{"parents":[{"commit":"b24698f3e23499303f171777441c81f9fc10222a","subject":"Merge \"feat: Support for custom prometheus requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b24698f3e23499303f171777441c81f9fc10222a"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 08:50:18.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fff81dc66982e8f2093ef2a08a5d7a9cabe83d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3fff81dc66982e8f2093ef2a08a5d7a9cabe83d4"}]},"branch":"refs/heads/master"},"d530421b25707dee502c67663262a27225b58c47":{"kind":"REWORK","_number":5,"created":"2022-10-13 09:21:53.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/5"}}},"commit":{"parents":[{"commit":"b24698f3e23499303f171777441c81f9fc10222a","subject":"Merge \"feat: Support for custom prometheus requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b24698f3e23499303f171777441c81f9fc10222a"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 09:18:16.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d530421b25707dee502c67663262a27225b58c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d530421b25707dee502c67663262a27225b58c47"}]},"branch":"refs/heads/master"},"2e0cd7cc23e5f4e5070bdb54b2cf80d473bcf715":{"kind":"REWORK","_number":6,"created":"2022-10-13 09:36:24.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/6"}}},"commit":{"parents":[{"commit":"b24698f3e23499303f171777441c81f9fc10222a","subject":"Merge \"feat: Support for custom prometheus requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b24698f3e23499303f171777441c81f9fc10222a"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 09:36:00.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2e0cd7cc23e5f4e5070bdb54b2cf80d473bcf715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2e0cd7cc23e5f4e5070bdb54b2cf80d473bcf715"}]},"branch":"refs/heads/master"},"4d5d808e86a7a47cddfea52df052c6321c27a1eb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-10-18 09:07:05.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/7"}}},"commit":{"parents":[{"commit":"a29d9dacd3f8443b9a756ad64b02368b107cb0b8","subject":"fix: update the click area of the id link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a29d9dacd3f8443b9a756ad64b02368b107cb0b8"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 08:02:35.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d5d808e86a7a47cddfea52df052c6321c27a1eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d5d808e86a7a47cddfea52df052c6321c27a1eb"}]},"branch":"refs/heads/master"},"8cad2821d490d7b56f545750878bc8cae3f1abb7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-10-18 09:28:36.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/8"}}},"commit":{"parents":[{"commit":"73e6153e240dc15d447345fdc613d58eae2009fa","subject":"fix: update the project name display in the rebuild instance action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73e6153e240dc15d447345fdc613d58eae2009fa"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:20:06.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8cad2821d490d7b56f545750878bc8cae3f1abb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8cad2821d490d7b56f545750878bc8cae3f1abb7"}]},"branch":"refs/heads/master"},"ba0d818c66809f40bf3edcd3975f63962032c5d3":{"kind":"REWORK","_number":9,"created":"2022-10-19 03:16:50.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/9"}}},"commit":{"parents":[{"commit":"73e6153e240dc15d447345fdc613d58eae2009fa","subject":"fix: update the project name display in the rebuild instance action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73e6153e240dc15d447345fdc613d58eae2009fa"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 03:16:02.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ba0d818c66809f40bf3edcd3975f63962032c5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ba0d818c66809f40bf3edcd3975f63962032c5d3"}]},"branch":"refs/heads/master"},"c9994f8097bcb8794bb9084bcc7418535a64be02":{"kind":"REWORK","_number":10,"created":"2022-10-19 04:03:05.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/10"}}},"commit":{"parents":[{"commit":"73e6153e240dc15d447345fdc613d58eae2009fa","subject":"fix: update the project name display in the rebuild instance action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73e6153e240dc15d447345fdc613d58eae2009fa"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 04:02:20.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: #1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c9994f8097bcb8794bb9084bcc7418535a64be02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c9994f8097bcb8794bb9084bcc7418535a64be02"}]},"branch":"refs/heads/master"},"a91c1f4926090013846ec12a27978a1fe70e1150":{"kind":"REWORK","_number":11,"created":"2022-10-19 04:12:56.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/11"}}},"commit":{"parents":[{"commit":"73e6153e240dc15d447345fdc613d58eae2009fa","subject":"fix: update the project name display in the rebuild instance action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73e6153e240dc15d447345fdc613d58eae2009fa"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 04:11:53.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: #1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a91c1f4926090013846ec12a27978a1fe70e1150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a91c1f4926090013846ec12a27978a1fe70e1150"}]},"branch":"refs/heads/master"},"3530e4fdb4db047021292cbba14de2d5973784c2":{"kind":"REWORK","_number":12,"created":"2022-10-19 08:07:27.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/12"}}},"commit":{"parents":[{"commit":"73e6153e240dc15d447345fdc613d58eae2009fa","subject":"fix: update the project name display in the rebuild instance action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73e6153e240dc15d447345fdc613d58eae2009fa"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:07:04.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: #1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3530e4fdb4db047021292cbba14de2d5973784c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3530e4fdb4db047021292cbba14de2d5973784c2"}]},"branch":"refs/heads/master"},"078293182ad8396e88db7e28985a60f968c8dba3":{"kind":"REWORK","_number":13,"created":"2022-10-19 08:28:47.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/13"}}},"commit":{"parents":[{"commit":"27fb2b2db04e00814abb9cf104354c589835a9a2","subject":"fix: fix the row auto selected when click the copy button edge","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/27fb2b2db04e00814abb9cf104354c589835a9a2"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:28:16.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: 1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/078293182ad8396e88db7e28985a60f968c8dba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/078293182ad8396e88db7e28985a60f968c8dba3"}]},"branch":"refs/heads/master"},"fdd6dbb5e38e63bad84f9dc68a13e2a18b3147be":{"kind":"REWORK","_number":14,"created":"2022-10-19 08:40:18.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/43/861143/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/861143/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/861143/14"}}},"commit":{"parents":[{"commit":"27fb2b2db04e00814abb9cf104354c589835a9a2","subject":"fix: fix the row auto selected when click the copy button edge","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/27fb2b2db04e00814abb9cf104354c589835a9a2"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-13 07:13:42.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:40:05.000000000","tz":480},"subject":"fix: Not show the data disk when creating vm by Instance snapshot","message":"fix: Not show the data disk when creating vm by Instance snapshot\n\nNot show the data disk when creating vm by Instance snapshot with system disk and data disk\n1. fetch instanceSnapshotDataVolumes data when change instance snapshot\n2. add Required Data Disk form item of BaseStep when create instance by snapshot\n3. show instance snapshot data disk\u0027s info if instance snapshot has system disk and data disk\n4. hide render default system disk when selecting Bootable Volume of Start Source\u0027s option\n\nCloses-Bug: #1992739\nChange-Id: I3a1daa8845cc9f73b8ea8d0216d49ad858b7168b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fdd6dbb5e38e63bad84f9dc68a13e2a18b3147be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fdd6dbb5e38e63bad84f9dc68a13e2a18b3147be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
