)]}'
{"id":"openstack%2Fskyline-console~861726","triplet_id":"openstack%2Fskyline-console~master~I035b1edb5e0441e63d7919ea5933731bd687c1d2","project":"openstack/skyline-console","branch":"master","topic":"bug-1992739","attention_set":{},"removed_from_attention_set":{"26304":{"account":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"last_update":"2022-10-19 01:23:17.000000000","reason":"removed on reply"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-10-19 12:07:58.000000000","reason":"Change was submitted"},"30356":{"account":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"last_update":"2022-10-19 01:24:18.000000000","reason":"removed on reply"},"34458":{"account":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"last_update":"2022-10-19 01:25:36.000000000","reason":"removed on reply"},"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-10-19 12:07:58.000000000","reason":"Change was submitted"},"35102":{"account":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"last_update":"2022-10-19 01:24:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I035b1edb5e0441e63d7919ea5933731bd687c1d2","subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","status":"MERGED","created":"2022-10-18 09:28:36.000000000","updated":"2022-10-26 01:26:07.000000000","submitted":"2022-10-19 12:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861726-bug-1992739","meta_rev_id":"8567067c5b4be8120adcd61608041f794e3636a3","_number":861726,"virtual_id_number":861726,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 12:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":2,"date":"2022-10-19 09:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":1,"date":"2022-10-19 09:11:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":1,"date":"2022-10-25 05:14:42.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":1,"date":"2022-10-26 01:26:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"value":1,"date":"2022-10-19 09:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 10:20:37.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-10-18 10:28:33.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-10-18 10:56:14.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"},{"updated":"2022-10-18 11:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 01:18:36.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-10-19 01:23:16.000000000","updated_by":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"reviewer":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"state":"REVIEWER"},{"updated":"2022-10-19 01:24:16.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2022-10-19 01:24:36.000000000","updated_by":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"reviewer":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"state":"REVIEWER"},{"updated":"2022-10-19 01:25:35.000000000","updated_by":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"reviewer":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"state":"REVIEWER"},{"updated":"2022-10-19 01:42:16.000000000","updated_by":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"reviewer":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"state":"REVIEWER"},{"updated":"2022-10-19 01:43:30.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-10-19 01:44:32.000000000","updated_by":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"reviewer":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"state":"REVIEWER"},{"updated":"2022-10-19 01:47:44.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-10-19 01:51:13.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2022-10-26 01:26:07.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"21c5123b521c64eccd7e74cb2c320e183b949116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-18 09:28:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"159b16e3dbc74b0e64357d3667578a1ada7c3142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-18 10:06:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e4f0d4eed31b610bde81c8eb01ba75789dd0b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-18 10:17:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f168277c497c205cced945fd3651b2b58675815","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-10-18 10:20:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a43d91267b7a6b258dfd2caab8e0397eecbf7aac","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-18 10:28:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc8ed66782d9514fd9a81d0896d787281bccfe74","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-18 10:56:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7172e2b7813b0a643658bec45dc0eecccd3fdc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:54:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39fa1966c5e74bfd9ec1d82f1a094b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab1ad11534a433980b1b46796c20972 : SUCCESS in 5m 54s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a15c01e25d8a48579247c72399afa237 : SUCCESS in 9m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b3a91510ef5a452cb670e987ee0d156d : SUCCESS in 7m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/88d0a29ded314d2f82246932ae41d75d : SUCCESS in 6m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/acdbfc8759f94d4da0991a8ca46ca7bb : SUCCESS in 1h 29m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/aee813f6ca854926b9b281225f2b31c8 : SUCCESS in 54m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5d4b72066f444e0aaa1ee2d18c76fcbd : SUCCESS in 48m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ca1499be0dc94e5baa95069ad22433c1 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f87f64fd46194dfe5a32590abbc994e677d26b76","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-19 01:18:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fc5050050b6d39ea922dc915f9e2c8e35c05c47b","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-10-19 01:23:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e1eae3825cc4c7319c3ecf45e7a47d72d34935a1","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-10-19 01:24:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3c922908961d07fe17bf5d78a80a6b03a4a9261d","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-10-19 01:24:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7ad12282ffad02bbea1cf157958fd8c382cd6771","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-10-19 01:25:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c87501d55f6fdc7e3ddabd9b1c35a49a2b8cc05c","author":{"_account_id":35158,"name":"weiwei.mo","email":"mo.weiwei@99cloud.net","username":"weiwei.mo"},"date":"2022-10-19 01:42:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5575458dbf86f1d9ca1acf9c1bfc15884cb016ce","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-19 01:43:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"def20ad7a9269ca9ff8d7bec57b351655ff557c9","author":{"_account_id":35191,"name":"liqi","email":"li.qi@99cloud.net"},"date":"2022-10-19 01:44:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6b6ad3c606a0538d87fd4aaff0452c1d4f038639","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-10-19 01:47:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"79a0bd2926fe77d529a1aa4cebfe4467835242e6","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-10-19 01:51:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4031f2805eff75d9c50895c65937d2739601c8be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 04:24:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fff7c6571b280477c37462bed45ef76f3a037d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 04:24:59.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/skyline-console for 861143,11","accounts_in_message":[],"_revision_number":4},{"id":"d08fc0babf3d0019210b360619f80bae9e551f0d","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2022-10-19 06:01:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5c54476a5812550fefb282ef1dfa85ffb44c5e6b","author":{"_account_id":35102,"name":"ZhangFei","email":"zhang.fei@99cloud.net","username":"zhang-fei"},"date":"2022-10-19 06:01:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5d8e7a65d9aefd2e0760031a78849b9d35e76c4e","author":{"_account_id":26304,"name":"chenhaooyan","email":"chen.haoyan@99cloud.net","username":"chenhaoyan"},"date":"2022-10-19 06:02:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b748664d01750d75f28c4845bece75a25cf38a0a","author":{"_account_id":34458,"name":"KanYiFei","email":"kan.yifei@99cloud.net","username":"kanyifei"},"date":"2022-10-19 06:02:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e794aea87e119ca0b12840c55370a4661d217ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 06:46:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9155731a4db85f9630ca66d9b0775a1b0b58fea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:00:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b28428301d476eaf913ab4f21b50af\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265bd2ce20ea49f993bf6831e8919698 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df025656bd8447c8a3fae86bb4458236 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fee1533241f6479886c62786a0f9dc45 : SUCCESS in 9m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4644f8588f654ab1b98d55908d1d8184 : SUCCESS in 4m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/06c321fd314e45c5a8f1d0aeee4bca2a : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3b8bc2f13e7f4cd8b985f53af7c9edd2 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":5},{"id":"dfefdb972c269024744bc5db816033c97e0f4c85","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-19 08:11:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a6302b1e09d6b400f99761d07cc6bbc02a129fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:22:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0bad4bb998c64ad50c0c5707ebf3540692f013bf","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-10-19 08:27:57.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e24fd32d868747aaec75ed1fef4f80f6dfc46b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:28:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ea17239f62aab3344ce5f564f6c184102789f13d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:42:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a32dfc32bdbecadae2c9f803843d31784f1c2222","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-19 08:43:55.000000000","message":"Topic set to bug-1992739","accounts_in_message":[],"_revision_number":8},{"id":"1ae5850ebf9390a1da42153219b6d3191ecaa72a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-19 09:05:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5537e87ce15bb1a2e6d439397207d45415e23e48","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-19 09:06:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"4285fd6f7b36fee3da1bc82b9cfd3c0515c9376f","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-19 09:11:49.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"9c0efb217f64ef4f74d708654aed6d2242979baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:48:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59159614a4c14ffea2e0357204f113ce\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/987f7da941ae4324a0185f8e465ff226 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea20d06905c148e5ab5bda102719ef8e : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/77e20974dc214b36abe1810b70f275fb : SUCCESS in 9m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ffb7b3bab57e48238fd69def9e9ccc91 : SUCCESS in 7m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3a74d8dc071e4cb98e3a11ea8ed57964 : SUCCESS in 5m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2ef6c35448ea493ebbca367be1294fc8 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":9},{"id":"d6458df00cf63202f538356cccdbf79f615f3079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:49:26.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"20f1e297dc8f1ba0eb82407d94e95999daa2f8f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:07:57.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8ead7ec4094365a281cbd0911a0444\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b97b023c865492e9b95d08e9dd1a552 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819e64f18f32477d9cae8b16dd39c3c7 : SUCCESS in 6m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9436c6e1df0140619ae42def931816d2 : SUCCESS in 10m 30s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/22180781ce25422590212e1f66da6f7a : SUCCESS in 5m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3ba0eb051197411f94c4e6b4bac17850 : SUCCESS in 6m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9e7a7eaed8d1402ba6c3ab58da1d50cd : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":9},{"id":"74f13841720d72177ed8c8076cd258ae3b52cf46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c59072ac680e05cc6fd0129b8a771fcc6eb3b0a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:09:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4db89f8264422d93f90a33c66a4b1e\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03ca1f60f89c4438aaca3f0b8c92cefa : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f859b1f137264806a133245f3005bb2c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":9},{"id":"c68bfd00da4c83f8c937800e1e3311b8a587a01a","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-25 05:14:42.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"8567067c5b4be8120adcd61608041f794e3636a3","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-10-26 01:26:07.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3b1ddc54a0b4c3633b6e87753012713cb2507a38","revisions":{"6d7610050aa7963b2366bc66cd68ab905b78078f":{"kind":"REWORK","_number":1,"created":"2022-10-18 09:28:36.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/1"}}},"commit":{"parents":[{"commit":"95eb964251d22a7e9e2279319ee7e0d0a5ab3998","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95eb964251d22a7e9e2279319ee7e0d0a5ab3998"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: 1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d7610050aa7963b2366bc66cd68ab905b78078f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6d7610050aa7963b2366bc66cd68ab905b78078f"}]},"branch":"refs/heads/master"},"6339bc2b7456148ae26fea36b6b9b66a731a7100":{"kind":"REWORK","_number":2,"created":"2022-10-18 10:06:19.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/2"}}},"commit":{"parents":[{"commit":"95eb964251d22a7e9e2279319ee7e0d0a5ab3998","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95eb964251d22a7e9e2279319ee7e0d0a5ab3998"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 10:05:54.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: 1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6339bc2b7456148ae26fea36b6b9b66a731a7100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6339bc2b7456148ae26fea36b6b9b66a731a7100"}]},"branch":"refs/heads/master"},"78dd7462f9951534bc2bcadd9cc5678a23bcc13a":{"kind":"REWORK","_number":3,"created":"2022-10-18 10:17:22.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/3"}}},"commit":{"parents":[{"commit":"95eb964251d22a7e9e2279319ee7e0d0a5ab3998","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95eb964251d22a7e9e2279319ee7e0d0a5ab3998"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 10:17:06.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: 1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78dd7462f9951534bc2bcadd9cc5678a23bcc13a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78dd7462f9951534bc2bcadd9cc5678a23bcc13a"}]},"branch":"refs/heads/master"},"91b30c717b75ef58ea01a03c8ae4be3fa1bb9a47":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-19 04:24:38.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/4"}}},"commit":{"parents":[{"commit":"95eb964251d22a7e9e2279319ee7e0d0a5ab3998","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/95eb964251d22a7e9e2279319ee7e0d0a5ab3998"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 04:23:27.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: #1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/91b30c717b75ef58ea01a03c8ae4be3fa1bb9a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/91b30c717b75ef58ea01a03c8ae4be3fa1bb9a47"}]},"branch":"refs/heads/master"},"d3f64bf2634a6f6108b2a27f5501df139732e1fa":{"kind":"REWORK","_number":5,"created":"2022-10-19 06:46:46.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/5"}}},"commit":{"parents":[{"commit":"6e23bbd82e54506c8e2b86b23ba35e164f05bb86","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6e23bbd82e54506c8e2b86b23ba35e164f05bb86"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 06:45:44.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: #1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d3f64bf2634a6f6108b2a27f5501df139732e1fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d3f64bf2634a6f6108b2a27f5501df139732e1fa"}]},"branch":"refs/heads/master"},"6f0acb7ec8357f65e91940c786c435c525f3205c":{"kind":"REWORK","_number":6,"created":"2022-10-19 08:22:18.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/6"}}},"commit":{"parents":[{"commit":"18acb0cc6089c5bf0b1018cfb1f75a9459ffd260","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/18acb0cc6089c5bf0b1018cfb1f75a9459ffd260"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:21:58.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: #1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f0acb7ec8357f65e91940c786c435c525f3205c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f0acb7ec8357f65e91940c786c435c525f3205c"}]},"branch":"refs/heads/master"},"2cd0d6088a8d1bf17dfb0709eb2b4ce08dff262c":{"kind":"REWORK","_number":7,"created":"2022-10-19 08:28:47.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/7"}}},"commit":{"parents":[{"commit":"6459bfeb727027b5669fd3c8352c6c12c4be6716","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6459bfeb727027b5669fd3c8352c6c12c4be6716"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:28:16.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: 1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2cd0d6088a8d1bf17dfb0709eb2b4ce08dff262c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2cd0d6088a8d1bf17dfb0709eb2b4ce08dff262c"}]},"branch":"refs/heads/master"},"e2771364dfaba0c6a9cfd752cac47f4234cdbc23":{"kind":"REWORK","_number":8,"created":"2022-10-19 08:42:31.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/26/861726/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/8"}}},"commit":{"parents":[{"commit":"33f90ae331fa0e6ddcd6afba193e31ce2168a953","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33f90ae331fa0e6ddcd6afba193e31ce2168a953"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-19 08:42:18.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota should be calculated according to the data disk of the selected instance snapshot and the newly added data disk\n\nCloses-Bug: #1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e2771364dfaba0c6a9cfd752cac47f4234cdbc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e2771364dfaba0c6a9cfd752cac47f4234cdbc23"}]},"branch":"refs/heads/master"},"3b1ddc54a0b4c3633b6e87753012713cb2507a38":{"kind":"REWORK","_number":9,"created":"2022-10-19 09:05:54.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/26/861726/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/861726/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/861726/9"}}},"commit":{"parents":[{"commit":"33f90ae331fa0e6ddcd6afba193e31ce2168a953","subject":"fix: No data disk is displayed on the ConfirmStep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33f90ae331fa0e6ddcd6afba193e31ce2168a953"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-18 09:21:12.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-19 09:05:05.000000000","tz":480},"subject":"fix: Calculate quota based on the data disk of the selected instance snapshot","message":"fix: Calculate quota based on the data disk of the selected instance snapshot\n\nWhen the Start source is the instance snapshot, the quota\nshould be calculated according to the data disk of the selected\ninstance snapshot and the newly added data disk\n\nCloses-Bug: #1992739\nChange-Id: I035b1edb5e0441e63d7919ea5933731bd687c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b1ddc54a0b4c3633b6e87753012713cb2507a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b1ddc54a0b4c3633b6e87753012713cb2507a38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
