)]}'
{"id":"openstack%2Fskyline-console~862859","triplet_id":"openstack%2Fskyline-console~master~Ib246c437e91814352d75fa2fc2bde9e7c1be8888","project":"openstack/skyline-console","branch":"master","topic":"feat-upload-children","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-10-28 10:07:49.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-10-28 08:26:28.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-10-28 10:07:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib246c437e91814352d75fa2fc2bde9e7c1be8888","subject":"feature: Support custom button for the upload","status":"MERGED","created":"2022-10-28 06:28:31.000000000","updated":"2022-10-31 01:06:40.000000000","submitted":"2022-10-28 10:07:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862859-feat-upload-children","meta_rev_id":"5c8852ad474a24057caeb5bae8c8cab927b8116e","_number":862859,"virtual_id_number":862859,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 10:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-10-28 08:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-10-28 06:44:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-10-28 07:13:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-31 01:06:40.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-10-28 08:26:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 06:44:44.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-10-28 07:13:50.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-10-28 07:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 08:19:43.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-10-28 08:19:43.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-10-31 01:06:40.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"93297495aeb5c4b556a80e968c4ee3134967e5df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-10-28 06:28:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a276c3bf25209131372cb0b679ac5c415857d3f","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-28 06:44:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66333033671f65cec5a71efc9dfa9ace3ed1ace5","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-28 07:13:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d04661d5caba2096e72ba7d520b0d308730dc64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 07:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5be63c910694c61a84f1a1b6e404eae\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b19df289cb5349ee8186d1941bde67d8 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8c30ad97884139925174e9fc420af9 : SUCCESS in 4m 38s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3ef49c48dfdf4744afb420241f4ae229 : SUCCESS in 8m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f59b3fb00dc7426aae236fedba9d7911 : SUCCESS in 5m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/334d4daf8daf4c7c942ed96cd7165ef5 : SUCCESS in 6m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/071770caae744a1596b00bf3327ac081 : SUCCESS in 1h 06m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/eb2d227867dc4ed99cdd55e678c36f3a : SUCCESS in 41m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/23f530770e8543819cbe52c7a423cea7 : SUCCESS in 43m 37s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/876ae1769481443ab9f55d7a41103eca : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fc65090cff275fc2a8b090948d0d72cbc2d21235","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-28 08:26:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abe00bbff79873d4fab3c82112d1aae4104dcb49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 08:27:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abeb2159afbd9a3cbfa3fa8b5b21ee2d6325f1bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 10:07:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ecbbce3bfb4c51900f94da9ccfa4b9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5918b826e5543b5a05bd6ef9954835d : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304b6169496843bd80af6d5cd63238c9 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2f59cdc02dc54e558b7e817d802fb199 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/74687418fd2949f882749a159a4c6509 : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/69aceffd2712423398ccebfe60c1045d : SUCCESS in 4m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/018baa2e39884d928b3caad2368edf30 : SUCCESS in 1h 39m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/63da01723a094d03aea9b1273fc073f4 : SUCCESS in 46m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/89467febfb6e43e18ac3cc5cfd226296 : SUCCESS in 41m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dccc21fb59a34a929f7a787f34a9395e : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"05d84aee50962fcb7c9fb08d7a81a1ee5de58055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 10:07:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5320573b420ffb9527ae6f1e3e6a7e79bb608935","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 10:09:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06043143be674d07bbe06dcefd16decc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac7bca71bc3248878a5ebf0374b6de0f : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baaff173680943cc831e1e41422d9d5f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"5c8852ad474a24057caeb5bae8c8cab927b8116e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-10-31 01:06:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"442d0557563ef4d5da3796348261f900481e64ef","revisions":{"442d0557563ef4d5da3796348261f900481e64ef":{"kind":"REWORK","_number":1,"created":"2022-10-28 06:28:31.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/59/862859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/862859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/862859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/862859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/862859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/862859/1"}}},"commit":{"parents":[{"commit":"21273908521e9b7b3cec67ff967d18e0d3647033","subject":"Merge \"feat: Support download file progress and abort download\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/21273908521e9b7b3cec67ff967d18e0d3647033"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-10-28 06:28:10.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-10-28 06:28:10.000000000","tz":480},"subject":"feature: Support custom button for the upload","message":"feature: Support custom button for the upload\n\nSupport custom button in the upload form item\n\nChange-Id: Ib246c437e91814352d75fa2fc2bde9e7c1be8888\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/442d0557563ef4d5da3796348261f900481e64ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/442d0557563ef4d5da3796348261f900481e64ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
