)]}'
{"id":"openstack%2Fskyline-console~862880","triplet_id":"openstack%2Fskyline-console~master~I6627a1fdbb92f95985c022ad15a19efabe4842c1","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-11-03 12:45:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6627a1fdbb92f95985c022ad15a19efabe4842c1","subject":"feat: add props in BaseForm","status":"MERGED","created":"2022-10-28 10:31:56.000000000","updated":"2022-11-07 02:23:00.000000000","submitted":"2022-11-03 12:45:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862880","meta_rev_id":"8aa9cffbf955e243fdc21a773b9024a8108fab16","_number":862880,"virtual_id_number":862880,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 12:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-03 09:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-11-03 10:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},{"value":1,"date":"2022-11-07 02:23:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-03 10:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 11:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 01:14:21.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-10-31 08:20:00.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-11-03 06:10:23.000000000","updated_by":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"reviewer":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"state":"REVIEWER"},{"updated":"2022-11-03 09:53:24.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-03 10:48:37.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"2932f46253acbed2038e8bbad6a5d663d2a611fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-10-28 10:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"520afde8797ea315c96e106ef0219e91a5553c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:53:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f0b4f8c2dc49d3965feba5ef4c6c18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b25840e519e473ca53718a54f9c251c : SUCCESS in 6m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/60827fb30f8448bc9aa453930dde57f5 : SUCCESS in 6m 33s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/707144a85d3d4c4787b4f803c95de8a4 : SUCCESS in 6m 16s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2f4a4b5ab5c84b9b8f724c4270363982 : SUCCESS in 6m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3c28eb26ca0f41f2a160a8da66216ccb : SUCCESS in 1h 18m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ce8c139c446a429bbe8b19d87db2c5e2 : SUCCESS in 44m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/22efcc2060ca4272aeba1d90b438c508 : SUCCESS in 47m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cd4d40d41bdc43faaade67a92c13490c : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e3106148e05fb77e5d46060565132480dbef7017","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-10-31 01:14:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e7fcdedfccfa262d91a7e1f6a61958f50466a0c","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-10-31 08:20:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c82d61130c6cdc5cc46a11d520060980928ca20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-03 01:58:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e31eff126d2f7f9575b90de1523b1403851e4dc","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-03 01:59:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d380c1b91e4a994c3c7f36ca60f219817c342a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 03:23:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3956949ec8d0459db92a8bbb835398ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d294c09d5ae4f7cb0a8f42e55fa2245 : SUCCESS in 5m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/10f5373697ec4471bad08838bb4ec909 : SUCCESS in 8m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d2487766cd3e4248bad282335af2bd30 : SUCCESS in 6m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6bda601e212d47a481d56015cf920882 : SUCCESS in 6m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a95a0ab45b6c48a7a38a168fff011f4b : SUCCESS in 1h 22m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a88cd139ee924e3c9c54123a49622a89 : SUCCESS in 1h 00m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3b490fc3296e45c3a4e9e241720e2764 : SUCCESS in 48m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5e4f511483544510a796b01583a77c06 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":2},{"id":"76f8b40bb2181b58e377d301d14a1970a7730a4c","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-03 05:57:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"55438778e0c5d8a6738d8cc931e106555a868635","author":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"date":"2022-11-03 06:10:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8fa54bf6088f64d6d297b653307f2389e67e5055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-03 06:12:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"491f13a0eadee4e6ebba1e100117181318ac37ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-03 06:38:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ef2142c076d503073b724d5bd732d7d3f1bf211f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 08:05:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e9ac6267814d8d8789d270b91412c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75aadd0b9782491a81fa6cddd2c1256b : SUCCESS in 3m 47s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d12164b4216846c1a41433f6149789a2 : SUCCESS in 7m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f0a9e0ab7ea24f6ea1e43135e6ac3ea9 : SUCCESS in 7m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2515572aa59f4c03945118219363ae99 : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bd8fad0c1f6f4ed781789e234c04672a : SUCCESS in 1h 16m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2ca7005722b84ec292f7b92949466397 : SUCCESS in 53m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3dd3ba40553d47c8b9804b64d784c6f3 : SUCCESS in 46m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2d22e17685d14bd296bf75810be5234c : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":4},{"id":"469418686591600a8fca64050f979c73a3c94f68","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-03 08:16:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cc9eaa8dfacdc4b5aac7b26c97bdbce0acd53cfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-03 09:25:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b4a705e21c6cb3939ecbb976dd3ad3e7404531f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-03 09:46:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6146922f3757dcb57fa96acdd427e714083fc747","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-03 09:53:24.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"da72939e5d4104dd8ffd6cc573f29b9f54f5b645","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-03 10:48:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"23ab154177d6432bc39080c48b0a9a992b6d1047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:22:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d59c47f22a734c6f8e4cb00241e68240\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516c2f2bcba240feae0fe14061838714 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9ab60b751a8c426e8ca04c23537c5320 : SUCCESS in 10m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d411faa782b54e8fb15d734c075a8e53 : SUCCESS in 7m 42s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1404f70cc2f74731b41fb561db1b2494 : SUCCESS in 5m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3a4416d64fef42079c741787f5e691f6 : SUCCESS in 1h 30m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d6410b748aa44c618fe4fcf7d73100bf : SUCCESS in 48m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/06efe932ddbe43da88606c6726606c3e : SUCCESS in 41m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b3b23363b35346b1a8e1361bf7c215e8 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":6},{"id":"60929e9e23b4fd3718faf4909c4afc225f277338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:23:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cf0274267fe5679e3ca2cfc3428a5983e4ad658f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:45:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70bd34195d0c4b78a3d11787efc79d06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b1396223c846569074d07788aadc3a : SUCCESS in 5m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/945700086b774e41a32fe4ddf4577910 : SUCCESS in 6m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bd62d83330654798b1f5ae5334a7b459 : SUCCESS in 7m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0d38ae4379764a138c028c1d3fe8dbcc : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e957da4436674c0496640cbf8fe955f5 : SUCCESS in 1h 21m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/025e804b681a434da4a545e6b11c53a8 : SUCCESS in 46m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/dd510250ddee4d5dbd6533ed6946c040 : SUCCESS in 49m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/020201ba8f614947903cdea461eaf9ec : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":6},{"id":"d8db0e3ac9077eefed92f54a4123262187fc9b15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:45:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"84a53e23ad857cd3d00f54a0b3db243d0c6ddde5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:46:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3038a45e4035453ea4599ddaab70686b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d04ea0c94d4a79a3700dd477973283 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6},{"id":"8aa9cffbf955e243fdc21a773b9024a8108fab16","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-07 02:23:00.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5a476a09a4791da932f2180c12480ba2ba7b2315","revisions":{"71ac32a99ffe321ce2c5656ef72f84561c2dc952":{"kind":"REWORK","_number":1,"created":"2022-10-28 10:31:56.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/1"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"subject":"feat: add hasFooter props in BaseForm","message":"feat: add hasFooter props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. support add placeholder props in the KeyValueInput component\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/71ac32a99ffe321ce2c5656ef72f84561c2dc952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/71ac32a99ffe321ce2c5656ef72f84561c2dc952"}]},"branch":"refs/heads/master"},"d1665fc062b3afdc4e51f6699e2ffaa6dacdc79a":{"kind":"REWORK","_number":2,"created":"2022-11-03 01:58:35.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/2"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-03 01:48:18.000000000","tz":480},"subject":"feat: add props in BaseForm","message":"feat: add props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. add setFormStyle props to set form style\n3. support add placeholder props in the KeyValueInput component\n4. add showConfirmAfterOk props to support show confirm after modal action in ActionButton\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d1665fc062b3afdc4e51f6699e2ffaa6dacdc79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d1665fc062b3afdc4e51f6699e2ffaa6dacdc79a"}]},"branch":"refs/heads/master"},"a9048a129438a002cedee7f861490a52e6059147":{"kind":"REWORK","_number":3,"created":"2022-11-03 06:12:58.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/3"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-03 06:12:32.000000000","tz":480},"subject":"feat: add props in BaseForm","message":"feat: add props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. add setFormStyle props to set form style\n3. support add placeholder props in the KeyValueInput component\n4. add showConfirmAfterOk props to support show confirm after modal action in ActionButton\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9048a129438a002cedee7f861490a52e6059147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9048a129438a002cedee7f861490a52e6059147"}]},"branch":"refs/heads/master"},"f3e9f08f05bbb748c58763f6ada3ed45955d0afa":{"kind":"REWORK","_number":4,"created":"2022-11-03 06:38:27.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/4"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-03 06:38:12.000000000","tz":480},"subject":"feat: add props in BaseForm","message":"feat: add props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. add setFormStyle props to set form style\n3. support add placeholder props in the KeyValueInput component\n4. add showConfirmAfterOk props to support show confirm after modal action in ActionButton\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3e9f08f05bbb748c58763f6ada3ed45955d0afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3e9f08f05bbb748c58763f6ada3ed45955d0afa"}]},"branch":"refs/heads/master"},"ace2e7b7b549084f332a718fbea5cb392245536d":{"kind":"REWORK","_number":5,"created":"2022-11-03 09:25:05.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/5"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-03 09:23:43.000000000","tz":480},"subject":"feat: add props in BaseForm","message":"feat: add props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. add formStyle and footerStyle props to set form and footer style\n3. support add placeholder props in the KeyValueInput component\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ace2e7b7b549084f332a718fbea5cb392245536d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ace2e7b7b549084f332a718fbea5cb392245536d"}]},"branch":"refs/heads/master"},"5a476a09a4791da932f2180c12480ba2ba7b2315":{"kind":"REWORK","_number":6,"created":"2022-11-03 09:46:06.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/80/862880/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/862880/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/862880/6"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-10-28 10:26:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-03 09:45:53.000000000","tz":480},"subject":"feat: add props in BaseForm","message":"feat: add props in BaseForm\n\n1. add hasFooter props to judge show the form footer in BaseForm\n2. add formStyle and footerStyle props to set form and footer style\n3. support add placeholder props in the KeyValueInput component\n\nChange-Id: I6627a1fdbb92f95985c022ad15a19efabe4842c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a476a09a4791da932f2180c12480ba2ba7b2315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5a476a09a4791da932f2180c12480ba2ba7b2315"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
