)]}'
{"id":"openstack%2Fskyline-console~863291","triplet_id":"openstack%2Fskyline-console~master~I92be0f8be8e66fa13763caab97e9968c2f956e68","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"35414":{"account":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"last_update":"2022-11-03 06:52:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92be0f8be8e66fa13763caab97e9968c2f956e68","subject":"Remove unicode literal from code","status":"MERGED","created":"2022-11-03 05:05:28.000000000","updated":"2022-11-03 06:53:41.000000000","submitted":"2022-11-03 06:52:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863291","meta_rev_id":"b25aaf1ded554e018834156a777483af49023dc6","_number":863291,"virtual_id_number":863291,"owner":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 06:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-03 06:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-03 06:51:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-03 05:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-03 06:24:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-03 06:35:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 05:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 05:57:39.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-11-03 06:24:43.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-11-03 06:35:51.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-03 06:51:57.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b3af0d6a17bd2e61c93a49cb5ca613ff3bcbc315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"date":"2022-11-03 05:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4ac7a5d0362ac3511bdef65715a5f4649992508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 05:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73eb0c2d81643fea3124ed3ba6cdd78\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/869934a183d74aa79e1904f1892bec7d : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e2a72c0dcb493f990a8765ed822382 : SUCCESS in 4m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c8c83cc3f87f4d34a23ded1a7a335842 : SUCCESS in 10m 19s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d300c2cb4904c32b744d15c1aeea94e : SUCCESS in 7m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fa29e65e43d242d5a1ac7587bc3acef2 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e8c0be3e45ecf51dc3839b21e653cd04c193882c","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-03 05:57:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"01e7d26739c7acddbb709d413af00a77d21c8304","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-03 06:24:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"48abc6cb9df8ba9a79fa85a6cccbabf2163d9f00","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-03 06:35:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8cd4f7951157fbef93dec7c2a48fa993ec87158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 06:36:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56190375c636aed4e3d9f1b39ed7e5d781bc4110","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-03 06:51:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1cc2c73838e828b86623c20114d782df633dcb38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 06:52:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54d9edaaead84bccbcb4a0729151faa2\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/046ffe94bd7b41c1a50af11766867c27 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2170cf8bf4984c388d89f0c97ce921d8 : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8ef5ed7ecb094b4c9a0769f6cb2da6dd : SUCCESS in 10m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8a52f5ee934e47af979a7f9baa3b1c46 : SUCCESS in 4m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5b1c1cc5e02444b8a717e6941a6131e7 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"74c6c12ce4cc852e2f9582495e715a52b7b1a676","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 06:52:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b25aaf1ded554e018834156a777483af49023dc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 06:53:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165e587220614ee6ae8c3d3da4c75598\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe846cf7f95469f8b361a1fc6638ace : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cfd4897f564ead81df2411409d8b84 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eaf68bea24fe4e60970a80b637a18ac1a4d5fa00","revisions":{"eaf68bea24fe4e60970a80b637a18ac1a4d5fa00":{"kind":"REWORK","_number":1,"created":"2022-11-03 05:05:28.000000000","uploader":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"ref":"refs/changes/91/863291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/91/863291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/863291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/863291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/91/863291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/91/863291/1"}}},"commit":{"parents":[{"commit":"b971ad860fbc80feb5929912e438adf1931aaf34","subject":"Merge \"feature: Support custom button for the upload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b971ad860fbc80feb5929912e438adf1931aaf34"}]}],"author":{"name":"lichenyu","email":"1798291787@qq.com","date":"2022-11-03 04:40:28.000000000","tz":480},"committer":{"name":"lcyuzzz","email":"1798291787@qq.com","date":"2022-11-03 04:46:26.000000000","tz":480},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nCloses bug: #1995548\n\nChange-Id: I92be0f8be8e66fa13763caab97e9968c2f956e68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eaf68bea24fe4e60970a80b637a18ac1a4d5fa00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eaf68bea24fe4e60970a80b637a18ac1a4d5fa00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
