)]}'
{"id":"openstack%2Fskyline-console~863292","triplet_id":"openstack%2Fskyline-console~master~I14a37c888b0ddb74581c0db257830c34290eb96c","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"35414":{"account":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"last_update":"2022-11-03 12:15:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14a37c888b0ddb74581c0db257830c34290eb96c","subject":"Remove unicode literal from code","status":"MERGED","created":"2022-11-03 07:27:44.000000000","updated":"2022-11-03 12:16:43.000000000","submitted":"2022-11-03 12:15:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863292","meta_rev_id":"c9c0aedc57a26005a28cbba1c4d998ac5ecd7823","_number":863292,"virtual_id_number":863292,"owner":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 12:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-03 10:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-03 08:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-03 08:00:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-03 10:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 08:00:11.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-11-03 08:16:03.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2022-11-03 08:56:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-03 10:49:04.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"05bb19d4308dcc827da2a1bb4751ad12540ddf80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"date":"2022-11-03 07:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"436c86bde41046fbeb55672918263b5e8ba7ecd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"date":"2022-11-03 07:40:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00fc7c8f1846a1b73b29e07f147ff98ab07552b2","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-03 08:00:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"35f9b664e543fa038a6d744a6e65199f638431e7","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-03 08:16:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b89ac18c79d1218593c0e8d966d143881998a15a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 08:56:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e2e4918bc04f0692110af6f6aad8e9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8db415a84b7a4186b82759d9ed6bc9f8 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54eea489c4746b6a6732b4ee0511846 : SUCCESS in 5m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dea8939d3e3242879e7337bdbcf434dc : SUCCESS in 7m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a977f35f20db4b9bb5ffcdfe468501ff : SUCCESS in 4m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bab14d4b0eca489dae6da35be8393193 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f09a7f6f2d7b4f9d96710cb61d6ce63d : SUCCESS in 1h 07m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4e040250c100445c80707ee60a451c5f : SUCCESS in 41m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1f6c4159b5a040eb9ca21686a3c5059a : SUCCESS in 41m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d9c218eec2534189ac61b475af57f329 : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":2},{"id":"deaa37ff2881569d0b0491675d08fb939223a836","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-03 10:49:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84849dce22c9c36bea7c9cb604a4af0d1b5ee5b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 10:49:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a48b4f36d08217242a764890eb292991e592337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:15:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d19e6b446e824bef9d20b1395209d6f3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7a7b473d1a4d15a19f1b8fcf6c5165 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5831b5a44b674dbd9f9db557a8221258 : SUCCESS in 5m 48s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0e146ff5e1fa44e199bd3c75cbe5baf5 : SUCCESS in 6m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b5aa3916e5a14f359e63fb848ed0c1bd : SUCCESS in 6m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d51980da1e3f430ba899290a04513ab7 : SUCCESS in 6m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/974b18e5b5754a8590a3b55836218553 : SUCCESS in 1h 24m 06s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bbaab260a87e4f248940f3d191bf1869 : SUCCESS in 43m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8b09937716d740fba0af936fc8d5476b : SUCCESS in 50m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/17a2c97893e24a82842684b5304642f3 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":2},{"id":"95ff1d87a47b7b153f5f7e4e165afecf5fa27b0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:15:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c9c0aedc57a26005a28cbba1c4d998ac5ecd7823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 12:16:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ba1ed17411408bb77a87ead4dc1ffd\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b53558296b0b481eb3501a99b157dc7b : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34462e8e38a048f19be2e7c5c8f45aaf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72eb092e860dbd5aae9b6d194714182d9ba48c55","revisions":{"9b178913475f8f9ddbce54876c1ac5c515238569":{"kind":"REWORK","_number":1,"created":"2022-11-03 07:27:44.000000000","uploader":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"ref":"refs/changes/92/863292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/863292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/1"}}},"commit":{"parents":[{"commit":"eaf68bea24fe4e60970a80b637a18ac1a4d5fa00","subject":"Remove unicode literal from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eaf68bea24fe4e60970a80b637a18ac1a4d5fa00"}]}],"author":{"name":"lcyuzzz","email":"1798291787@qq.com","date":"2022-11-03 07:26:14.000000000","tz":480},"committer":{"name":"lcyuzzz","email":"1798291787@qq.com","date":"2022-11-03 07:27:21.000000000","tz":480},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nCloses bug: #1995548\n\nChange-Id: I14a37c888b0ddb74581c0db257830c34290eb96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b178913475f8f9ddbce54876c1ac5c515238569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b178913475f8f9ddbce54876c1ac5c515238569"}]},"branch":"refs/heads/master"},"72eb092e860dbd5aae9b6d194714182d9ba48c55":{"kind":"REWORK","_number":2,"created":"2022-11-03 07:40:29.000000000","uploader":{"_account_id":35414,"name":"lichenyu","email":"1798291787@qq.com","username":"lcyuzzz"},"ref":"refs/changes/92/863292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/92/863292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/92/863292/2"}}},"commit":{"parents":[{"commit":"eaf68bea24fe4e60970a80b637a18ac1a4d5fa00","subject":"Remove unicode literal from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eaf68bea24fe4e60970a80b637a18ac1a4d5fa00"}]}],"author":{"name":"lcyuzzz","email":"1798291787@qq.com","date":"2022-11-03 07:26:14.000000000","tz":480},"committer":{"name":"lcyuzzz","email":"1798291787@qq.com","date":"2022-11-03 07:40:19.000000000","tz":480},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nCloses bug: #1995548\n\nChange-Id: I14a37c888b0ddb74581c0db257830c34290eb96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/72eb092e860dbd5aae9b6d194714182d9ba48c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/72eb092e860dbd5aae9b6d194714182d9ba48c55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
