)]}'
{"id":"openstack%2Fskyline-console~863815","triplet_id":"openstack%2Fskyline-console~master~I588af877fe776dbf7d44b11ad350566bf231121f","project":"openstack/skyline-console","branch":"master","topic":"magnum-redesign","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-08 01:47:31.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-08 01:42:45.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-11-08 02:30:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I588af877fe776dbf7d44b11ad350566bf231121f","subject":"feature: Optimization of Magnum","status":"MERGED","created":"2022-11-07 08:22:01.000000000","updated":"2022-11-15 02:14:07.000000000","submitted":"2022-11-08 02:30:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863815-magnum-redesign","meta_rev_id":"a6bc6acfadb887142389a7637bea22be39df069a","_number":863815,"virtual_id_number":863815,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 02:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-08 01:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-11-08 01:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-15 02:14:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-08 00:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-08 01:47:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 08:24:15.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-11-07 09:05:19.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-07 09:05:19.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-07 09:25:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 02:14:07.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"}],"messages":[{"id":"2ab1ec0861694f0bbf18a52ebceae7460583e8e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-07 08:22:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5712fe19b63ba44dee561629c23fa02342b3ce4","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-07 08:24:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a0d9174f782c2557ba666581f1537208b7a5d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 09:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abff7c7c171423699f121ef2cbefcb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a653e1d1c447b1a8ec2937e57bf79f : SUCCESS in 5m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4feca42c2ee5405e9c91050bcea7a7c8 : SUCCESS in 8m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c856ea1e106549caa18962b4b6733549 : SUCCESS in 6m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2fe7d5ddcb5a45f1949e6c1a799e0e2c : SUCCESS in 5m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/46dbf0f8a4d64f0193a9fcbd4a071499 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1e60221680527aa456665c1f1150e656acfe59ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-07 09:44:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9931ac5f434163eb06bb3385ce8c20c4e4867fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:27:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3c4c12941c4ed99b06b60b1338f7f2\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4b62316a2de40ee9ef77d90e2bcfb46 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70082eb251e4a40b0bfb10b04d2dc73 : SUCCESS in 4m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6be76911b53c444c950412ec506d650b : SUCCESS in 6m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b0d72c883f3e488ea641b7ce1ffb5584 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e2d22fb0fdba417fa170e2f52cdf6fb5 : SUCCESS in 4m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/61b3102a5ee24230bae34433ca596958 : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6eabbd50a1bebf0ce4bb8482af07e961ff6b9e38","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-08 00:51:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d4856e92ad81353d58fdc8eee0920f1583519004","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-08 01:42:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd1cdcf9bd7606b9f93ba750c1e04061aa6eb14d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-08 01:47:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1b2e82b8f6c918df124dfa2a13cee615c4703664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 01:48:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55bb234c129601e0f0778bd15b1b5f9ddb86ebf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 02:30:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e574d1338b7343c0b32e1a674563ab1b\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ade4c7ed30e4eb581becfc63d93efa0 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a91028f047472aa604b712d051e66b : SUCCESS in 4m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7635d89bac3d42da95a6cb18355d1b51 : SUCCESS in 10m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/807a4db80ac847f9b96b3830b24a6055 : SUCCESS in 4m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a66d6ee3fb0242f1baddac6ad93c9a0f : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3def611c83ab4734bf915ea130828fcb : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f3fd6856089e9a8ac862bcd242c78814dc44eec9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 02:30:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5c6a4a5335585510c8663de1613f54c0668afee5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 02:32:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38bcbc011dda49e3ae2bebcce711d6eb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77d9203287744b83912e02e2f61be966 : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249b079b31c3464daacbb6549ca81673 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a6bc6acfadb887142389a7637bea22be39df069a","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-15 02:14:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","revisions":{"813e6312e776dbe1a2db9861e9fe388793b5554a":{"kind":"REWORK","_number":1,"created":"2022-11-07 08:22:01.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/15/863815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/863815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/1"}}},"commit":{"parents":[{"commit":"f73d5a37bf3279aa40d78a7611c301b2ff5f58f1","subject":"Merge \"feat: add props in BaseForm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f73d5a37bf3279aa40d78a7611c301b2ff5f58f1"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-07 08:21:19.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-07 08:21:19.000000000","tz":480},"subject":"feature: Optimization of Magnum","message":"feature: Optimization of Magnum\n\n1. Update some form item labels for better semantics\n2. Make the template name be required when create the template\n3. Support create cluster instance in template list\n\nChange-Id: I588af877fe776dbf7d44b11ad350566bf231121f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/813e6312e776dbe1a2db9861e9fe388793b5554a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/813e6312e776dbe1a2db9861e9fe388793b5554a"}]},"branch":"refs/heads/master"},"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f":{"kind":"REWORK","_number":2,"created":"2022-11-07 09:44:22.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/15/863815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/863815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/863815/2"}}},"commit":{"parents":[{"commit":"f73d5a37bf3279aa40d78a7611c301b2ff5f58f1","subject":"Merge \"feat: add props in BaseForm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f73d5a37bf3279aa40d78a7611c301b2ff5f58f1"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-07 08:21:19.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-07 09:43:50.000000000","tz":480},"subject":"feature: Optimization of Magnum","message":"feature: Optimization of Magnum\n\n1. Update some form item labels for better semantics\n2. Make the template name be required when create the template\n3. Support create cluster instance from template list\n\nChange-Id: I588af877fe776dbf7d44b11ad350566bf231121f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
