)]}'
{"id":"openstack%2Fskyline-console~863983","triplet_id":"openstack%2Fskyline-console~master~Iee64af2821bd034e3818166308512d946e44cbe7","project":"openstack/skyline-console","branch":"master","topic":"global-nav","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-09 10:28:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee64af2821bd034e3818166308512d946e44cbe7","subject":"feat: add global navigation","status":"MERGED","created":"2022-11-08 10:00:03.000000000","updated":"2022-11-15 02:14:01.000000000","submitted":"2022-11-09 10:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863983-global-nav","meta_rev_id":"3ea451b41f150f8e4cb4060e93417eb6c552a1cc","_number":863983,"virtual_id_number":863983,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-09 10:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-09 08:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-15 02:14:01.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-09 04:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":1,"date":"2022-11-09 07:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-09 10:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-09 08:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 11:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 04:43:27.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-11-09 07:21:08.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-11-09 08:50:30.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-09 10:27:00.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"},{"updated":"2022-11-15 02:14:01.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"}],"messages":[{"id":"812a8587c724f1fa325682335202d8b620a5a27e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-08 10:00:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c31b23ab6369df67dea5140a95fd03b140f4b0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:24:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/324ce749e3314b9a8e1a8f70e26c6a77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25539701e2c4227962faff22bcbd8aa : SUCCESS in 5m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/839c312f396247de80bfaf894ea5cc7d : SUCCESS in 8m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4f1ab547c69943db85ef159aef02fe82 : SUCCESS in 6m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/40d0daf891594e279e0b5a785b01c025 : SUCCESS in 6m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/40ade6db6d6d48c8bf0844a7bee8c83d : SUCCESS in 1h 23m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b7f0a282396246cc95c3145876c4156a : SUCCESS in 52m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e48051bb3bd44f58be2554bfef50822c : SUCCESS in 49m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d2416b98424c46ac9192f30dd84f8bd4 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c1edc47f73e10b00987afbb56299c7de8e551a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-09 01:57:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a2c325eb0595de5b53bfbca157034998c64b73d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 03:30:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a6a2dd01b6d49b1aabbcda72eab34f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9161190d62344251a37f6eff5aa5255c : SUCCESS in 4m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0dfdce8445734b9f8af2d2aaacf0a5c6 : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bdb61bab10ce4fe699b8923d0cdfa504 : SUCCESS in 8m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7311c8f822914f779b60c56e5112e0f9 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2205332e300749bca46a5e66e4756f83 : SUCCESS in 1h 31m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c2e1f876f01d45c285ca1206471c87f1 : SUCCESS in 1h 20m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d8f116887eb44bed830020f5e5b81863 : SUCCESS in 58m 44s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/18c6988befbe413585172d0e1454d87f : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":2},{"id":"973340df6692c535d3324188542d5c395d3420ba","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-09 04:43:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"893109a8a6d775ce696c5a586a7929c87e5eb4d4","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 07:21:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1267a1f25ad1f1571d87e14be4c378c494afc180","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-09 08:50:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7984a9acf5a0cb28fba6a28b7016aa5ac9c26ff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 08:51:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e881be6cd6b63e13de085002a8aefeaf7b2d4da2","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-11-09 10:27:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a640e320cef3f347f303d07c5e58b9bef1106b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:28:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fbc4bd8f8944254a3bea414b5d0e03a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2554aeee2b064680a103e0ead63615a7 : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4463065b61cd495d9affcef3780f9537 : SUCCESS in 8m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/61dabd97fca349e8875f1c0e6523a1fa : SUCCESS in 7m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5e5877b370f6469095fe022e80b528f2 : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6f3d3637d2bb4cf29a54cd339983fe4a : SUCCESS in 1h 32m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/43cc8e320a8e437a956dbe66d60500b1 : SUCCESS in 54m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d7acd3d1888d4e018fffb79bd3e509a4 : SUCCESS in 49m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/78fb9f2293874b9a9c4bcaf9f56f45aa : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":2},{"id":"120601d4df1f16a39d07b437f7b1c0b5b3b2da3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:28:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"866da240ef74f9517f7b9faf9ad31d3d14dfb753","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:29:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d7c488ac26442893cad0955da2fddc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32f468c18684e869f579bc100d39011 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"3ea451b41f150f8e4cb4060e93417eb6c552a1cc","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-15 02:14:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"985c03af9565a09701e5f49414fe326ce54705d2","revisions":{"df035ff2d36a5602847e9c9685a35cc6707e04e8":{"kind":"REWORK","_number":1,"created":"2022-11-08 10:00:03.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/83/863983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/83/863983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/1"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-08 09:49:10.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-08 09:58:36.000000000","tz":480},"subject":"feat: add global navigation","message":"feat: add global navigation\n\n1. Add global navigation at the left top position of the whole page\n2. Update the right menu layout to adjust the global navigation\n3. The global navigation show the extended menu in the console/user center/administrator\n\nChange-Id: Iee64af2821bd034e3818166308512d946e44cbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df035ff2d36a5602847e9c9685a35cc6707e04e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df035ff2d36a5602847e9c9685a35cc6707e04e8"}]},"branch":"refs/heads/master"},"985c03af9565a09701e5f49414fe326ce54705d2":{"kind":"REWORK","_number":2,"created":"2022-11-09 01:57:52.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/83/863983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/83/863983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/83/863983/2"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-08 09:49:10.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-09 01:57:38.000000000","tz":480},"subject":"feat: add global navigation","message":"feat: add global navigation\n\n1. Add global navigation at the left top position of the whole page\n2. Update the right menu layout to adjust the global navigation\n3. The global navigation show the extended menu in the console/user center/administrator\n\nChange-Id: Iee64af2821bd034e3818166308512d946e44cbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/985c03af9565a09701e5f49414fe326ce54705d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/985c03af9565a09701e5f49414fe326ce54705d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
