)]}'
{"id":"openstack%2Fskyline-console~864079","triplet_id":"openstack%2Fskyline-console~master~I567db87e369264d9b1d1372332868267b478cf46","project":"openstack/skyline-console","branch":"master","topic":"global-nav","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-09 10:28:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I567db87e369264d9b1d1372332868267b478cf46","subject":"feat: update menu in the layout","status":"MERGED","created":"2022-11-09 04:30:53.000000000","updated":"2022-11-15 02:13:53.000000000","submitted":"2022-11-09 10:28:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864079-global-nav","meta_rev_id":"3ada1e217457805fe5bef2618af162d21c52d8d9","_number":864079,"virtual_id_number":864079,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-09 10:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-09 08:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-15 02:13:53.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-09 04:43:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":1,"date":"2022-11-09 07:21:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-09 08:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 04:43:18.000000000","updated_by":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"reviewer":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"state":"REVIEWER"},{"updated":"2022-11-09 05:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 07:21:14.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-11-09 08:51:27.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-15 02:13:53.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"}],"messages":[{"id":"ee7d3b47d6900f984274728cd9343fabb12bb4a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-09 04:30:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b1bb91e371f59ec9946c918564a8ef6ace7c1d2","author":{"_account_id":35161,"name":"syw","email":"sun.youwei@99cloud.net","username":"syw"},"date":"2022-11-09 04:43:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d243dde6e147611c71276d6277fe9fb7ccc3cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 05:38:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b18ae35548140fe84a2759ec8916212\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da815dd4ad345a3a4a61604e648636c : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/41aa47125b0941c19ee9d469c39769d5 : SUCCESS in 6m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f1fd190e041d408a98d42ef9dab5da1a : SUCCESS in 4m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/72ad21c39c964ea59169c71a36dff6f6 : SUCCESS in 4m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/516a1e7918e947b690d2341796c4cdad : SUCCESS in 1h 05m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c0247dae39374288b4d264576558671e : SUCCESS in 1h 00m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/40789d9f16d9431ebecff7bb31fe42d7 : SUCCESS in 37m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2aab5cf6b59044009f19c471633a02b0 : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f7283059da67a676364b6ef24198e7e1bff7146c","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 07:21:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ad314ec6b3ab11e33a29c2ef30c0c8137d3eb10","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-09 08:51:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0de3a32696efc5e2f787f629829ec34e9f7048c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 08:52:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c58660dc9b3986cf0c1810796f4dce79823e0ee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:28:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbcb9d9da5ce4a2db4b4be04d026a5e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8339bb80df684a8d8c2e6f35619d41eb : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fedda3b37b424d9dbb1996eed3cd353f : SUCCESS in 8m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/793616e6bc96473e8682e12ea7626157 : SUCCESS in 7m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a7a8d9da57b1474b8087d08468fb556f : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e14c9654f5dd41d3a08574deb83df6f8 : SUCCESS in 1h 29m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2ebdec8de1164710bab088798957cbff : SUCCESS in 1h 13m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/85fd8cdd88ed4e21b790398b7791d419 : SUCCESS in 43m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/68baa84525ff40ef88a803bff062ee9c : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":1},{"id":"24abb294b8ffc0e9d42158c270625411a0ee5d59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:28:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62c3a21bb0c4b5bcf20abe5ae72242fee46d0537","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:30:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865d5a0f50784fcfb53f6deaf517dc3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1956260cd64351837602136fefbc33 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1},{"id":"3ada1e217457805fe5bef2618af162d21c52d8d9","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-15 02:13:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3543ed1b50f7f9661e29ce2f5882b1c55fe71a4","revisions":{"e3543ed1b50f7f9661e29ce2f5882b1c55fe71a4":{"kind":"REWORK","_number":1,"created":"2022-11-09 04:30:53.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/79/864079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/79/864079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/864079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/864079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/79/864079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/79/864079/1"}}},"commit":{"parents":[{"commit":"985c03af9565a09701e5f49414fe326ce54705d2","subject":"feat: add global navigation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/985c03af9565a09701e5f49414fe326ce54705d2"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-09 04:30:28.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-09 04:30:28.000000000","tz":480},"subject":"feat: update menu in the layout","message":"feat: update menu in the layout\n\n1. Update the menu style to adjust the global nav display\n2. Update the menu open keys when the url change: the menu will auto open the current url\u0027s father menu\n\nChange-Id: I567db87e369264d9b1d1372332868267b478cf46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3543ed1b50f7f9661e29ce2f5882b1c55fe71a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3543ed1b50f7f9661e29ce2f5882b1c55fe71a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
