)]}'
{"id":"openstack%2Fskyline-console~864097","triplet_id":"openstack%2Fskyline-console~master~I5a698ccde7454f8d34e6d0d3ed726582708855b0","project":"openstack/skyline-console","branch":"master","topic":"magnum-redesign","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-09 10:12:33.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-09 10:00:08.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-11-09 11:57:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a698ccde7454f8d34e6d0d3ed726582708855b0","subject":"Fix the cluster and template of magnum","status":"MERGED","created":"2022-11-09 09:38:34.000000000","updated":"2022-11-15 02:13:39.000000000","submitted":"2022-11-09 11:57:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864097-magnum-redesign","meta_rev_id":"4391be0d1ba8276a9bc1a3e40910ede461c6b722","_number":864097,"virtual_id_number":864097,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-09 11:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-09 10:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-11-09 10:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-15 02:13:39.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-09 10:12:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 09:43:07.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-09 09:43:07.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-09 10:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 02:13:39.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"}],"messages":[{"id":"969455f0b39c7d5f825d4cc5e6dbe4ee9b9c77c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 09:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49209826ff4d2b32bdb517e948d725632c5dc226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 09:42:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"150cae69d74eef5b96d24990635cec6385cb2ec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 09:43:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"60faa88c6fbeb6a0f71849748dadf4bb8fe27b5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-09 09:57:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"489564d7df30498afa0ca62ad363ff50efe0e7e2","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-09 10:00:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9cd8f0e1b539af95fb7a688239db03d20dd718cc","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-09 10:12:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3cd3e1af287d33bcfd89c8afd163cd4ecb0fd1e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:48:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cba1760eb614f1ca5b352a9cdc90c14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e86158a5fdf4fda9812fc601fa8b93c : SUCCESS in 4m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0b811cca4a954491a35c51476aa28d7c : SUCCESS in 6m 18s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bc5842966be44eab86f85162e902e4cd : SUCCESS in 4m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e49e301c32ec4837b8bff1e42d7b5e6c : SUCCESS in 5m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4f3c86a9de6344eebac44f6a4ee61fe7 : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":4},{"id":"80179d3cbe8e34eeef87d78c81eca3c43870cca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:49:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c1d7025c2bb4bd9e35b2ba25cc8a7886a9dc7f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 11:57:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cee95f3e3624b22947586eb630b23bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b94be8d7cc54bec854307651644e109 : SUCCESS in 5m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b05bfed9f8a44cb195ce62bdd7aabe02 : SUCCESS in 8m 43s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2466a59f368448cda5b5d0ef5b9e62fe : SUCCESS in 9m 58s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7298f047f0264beabcbe648b7d9ac059 : SUCCESS in 6m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c9efe56dd08b449f9f1242b043bcac0d : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":4},{"id":"29c39f6a0e5b4ab12f01e4967d23e1c962c39559","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 11:57:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8c4858f9309ecfc63a362a03bd85c336fd3b9d82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 11:58:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/593794473c1c44a0b497ffbc06d27811\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e485f225847c4be5b895637a94397555 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4},{"id":"4391be0d1ba8276a9bc1a3e40910ede461c6b722","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-15 02:13:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5eab747fb30f1680cdba3d41dac440e1bf331c6d","revisions":{"1c9fd1570ed84575dfe5273504be71a02b0469dd":{"kind":"REWORK","_number":1,"created":"2022-11-09 09:38:34.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/97/864097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/864097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/1"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:36:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:36:52.000000000","tz":480},"subject":"Fix the cluster and template of magnum Edit","message":"Fix the cluster and template of magnum Edit\n\n1. When creating a cluster instance, the parameters are the same as the template, change them to optional\n2. Fix some tips of form item to get a better understanding\n3. Change the filter for image list by zed spec of openstack magnum user guide document\n\nCloses-Bug: #1996046\nChange-Id: I5a698ccde7454f8d34e6d0d3ed726582708855b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c9fd1570ed84575dfe5273504be71a02b0469dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1c9fd1570ed84575dfe5273504be71a02b0469dd"}]},"branch":"refs/heads/master"},"a05e40d6c852885d043d324ced81cf45a6b18d0d":{"kind":"REWORK","_number":2,"created":"2022-11-09 09:42:25.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/97/864097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/864097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/2"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:36:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:42:10.000000000","tz":480},"subject":"Fix the cluster and template of magnum Edit","message":"Fix the cluster and template of magnum Edit\n\n1. When creating a cluster instance, the parameters are the same as the template, change them to optional\n2. Fix some tips of form item to get a better understanding\n3. Change the filter for image list by zed spec of openstack magnum user guide document\n\nCloses-Bug: #1996046\nChange-Id: I5a698ccde7454f8d34e6d0d3ed726582708855b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a05e40d6c852885d043d324ced81cf45a6b18d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a05e40d6c852885d043d324ced81cf45a6b18d0d"}]},"branch":"refs/heads/master"},"1db6c8cc08660880315646ad5f07cd252374f486":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-09 09:43:20.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/97/864097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/864097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/3"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:36:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:43:20.000000000","tz":0},"subject":"Fix the cluster and template of magnum","message":"Fix the cluster and template of magnum\n\n1. When creating a cluster instance, the parameters are the same as the template, change them to optional\n2. Fix some tips of form item to get a better understanding\n3. Change the filter for image list by zed spec of openstack magnum user guide document\n\nCloses-Bug: #1996046\nChange-Id: I5a698ccde7454f8d34e6d0d3ed726582708855b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1db6c8cc08660880315646ad5f07cd252374f486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1db6c8cc08660880315646ad5f07cd252374f486"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5eab747fb30f1680cdba3d41dac440e1bf331c6d":{"kind":"REWORK","_number":4,"created":"2022-11-09 09:57:21.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/97/864097/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/97/864097/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/97/864097/4"}}},"commit":{"parents":[{"commit":"5ba32d6ddea61765934ebbf0dfa1447bee0ab24f","subject":"feature: Optimization of Magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5ba32d6ddea61765934ebbf0dfa1447bee0ab24f"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:36:52.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-09 09:56:59.000000000","tz":480},"subject":"Fix the cluster and template of magnum","message":"Fix the cluster and template of magnum\n\n1. When creating a cluster instance, the parameters are the same as the template, change them to optional\n2. Fix some tips of form item to get a better understanding\n3. Change the filter for image list by zed spec of openstack magnum user guide document\n\nCloses-Bug: #1996046\nChange-Id: I5a698ccde7454f8d34e6d0d3ed726582708855b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5eab747fb30f1680cdba3d41dac440e1bf331c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5eab747fb30f1680cdba3d41dac440e1bf331c6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
