)]}'
{"id":"openstack%2Fskyline-console~864249","triplet_id":"openstack%2Fskyline-console~master~Ie9f12f2d6747429966e62bbc0bbd72d9686864cd","project":"openstack/skyline-console","branch":"master","topic":"global-nav","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-11 10:41:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9f12f2d6747429966e62bbc0bbd72d9686864cd","subject":"feat: update the click area of links in the global navigation","status":"MERGED","created":"2022-11-11 07:35:05.000000000","updated":"2022-11-15 02:13:11.000000000","submitted":"2022-11-11 10:41:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864249-global-nav","meta_rev_id":"451ada4bffed2eada0eb0d0d7e8357a4b2e03f6d","_number":864249,"virtual_id_number":864249,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-11 10:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-11 09:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-11-15 02:13:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":1,"date":"2022-11-11 09:10:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-11 09:15:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-11 09:00:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-11 09:10:54.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-11-11 09:15:48.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-15 02:13:11.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"}],"messages":[{"id":"ee2ed5d67ca85a3c14e12f2fe0977e74cc25272e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-11 07:35:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"679b49a9c6d1087c111d9207ade62972f0d150c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 09:00:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2cc670ce294499a3d0d521df68d37f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b8c834e1454dbea2ec99cdcbbd32e2 : SUCCESS in 4m 48s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/78e4d2135fdc44d4a6519360e64d5634 : SUCCESS in 9m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f6444b0067be43ce9d3f58f0748d4613 : SUCCESS in 4m 55s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/776e1fa744514c65999bd549e1d8990c : SUCCESS in 4m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/da9dd8d0dc28494ca6e9c3b683a93769 : SUCCESS in 1h 23m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/79d207af88a74c188b54c87a2c50c374 : SUCCESS in 42m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a2c8087d12b8433eb625d052cb66d791 : SUCCESS in 53m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e34429adaa084a5c97c0b42e35d45605 : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6928108f71f71b5addaca857d664772184281132","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-11 09:10:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05790a052a1a6f7042d3966f18da9afd014c02ff","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-11 09:15:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c50a31373d9614541158bdd686de51ca6f4de8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 09:16:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76e1bb5bf799f7f4cf0cc705960251987ab6b6e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db758bc6b79942a39356beb3df122344\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d8d8eaec79483ba13b9602558de4e3 : SUCCESS in 5m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/93f706f51b094557ad4dcb4ae1d76812 : SUCCESS in 8m 12s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8436f31fb68e44b6b23ac71fb263299c : SUCCESS in 6m 15s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5d538e18b5fe473f9ebf09f48d720b29 : SUCCESS in 6m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dc4c38fb20474cbca80b8adf7ba684d2 : SUCCESS in 1h 24m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a611fd6895d0479388c16c64a479b8f7 : SUCCESS in 44m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d512b77d2f5940dcaeb0b6b196ba90d6 : SUCCESS in 47m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1019fea5f0ee4d7fb323aeaec2ad1e2e : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":1},{"id":"519ec0c142b0bda50f4ebc6775412ed87cc53137","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:41:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"778c832e2bb32ec5822c7d9beadfb84ac8c64298","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:42:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a06a4b0a8fa40bc90abee0ba571984d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b373a78b75394bfb98b9a34fa3f98e88 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"451ada4bffed2eada0eb0d0d7e8357a4b2e03f6d","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-11-15 02:13:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e30ad69229938c89e213ac91af3ca682ee17252","revisions":{"7e30ad69229938c89e213ac91af3ca682ee17252":{"kind":"REWORK","_number":1,"created":"2022-11-11 07:35:05.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/49/864249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/49/864249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/864249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/864249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/49/864249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/49/864249/1"}}},"commit":{"parents":[{"commit":"f89ffe6149f83fa43e924d5c3821579dba5d2419","subject":"Fix the state and default value when create cluster instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f89ffe6149f83fa43e924d5c3821579dba5d2419"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-11 07:34:38.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-11-11 07:34:38.000000000","tz":480},"subject":"feat: update the click area of links in the global navigation","message":"feat: update the click area of links in the global navigation\n\n1. Update the click area of the links in the global navigation\n2. Update the proptype of the globalNav component\n\nChange-Id: Ie9f12f2d6747429966e62bbc0bbd72d9686864cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7e30ad69229938c89e213ac91af3ca682ee17252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7e30ad69229938c89e213ac91af3ca682ee17252"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
