)]}'
{"id":"openstack%2Fskyline-console~864251","triplet_id":"openstack%2Fskyline-console~master~Ib3f33ad45b5630b1b7d56573399873decb8a0a70","project":"openstack/skyline-console","branch":"master","topic":"fix-i18n-monitor","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-11 10:17:01.000000000","reason":"Change was abandoned"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-11 10:17:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib3f33ad45b5630b1b7d56573399873decb8a0a70","subject":"feature: Add translation for monitor center","status":"ABANDONED","created":"2022-11-11 08:53:02.000000000","updated":"2022-11-11 10:17:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9cd1cfe96c5a88aace061b3d9c4cdd51e7e3583b","_number":864251,"virtual_id_number":864251,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-11 10:03:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-11 09:33:20.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-11 09:33:20.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-11 10:03:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27d194f49db8bab3d578a404a007a6b5e0525fd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-11 08:53:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b178c7e73fd7a218b218bd4b0a458a75b19fd4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 10:03:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b8f1200a3e454a937833ea96c529b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4270ca1470644f85bc745cfa5b7ac98c : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b380e3fb409649b99f511c282ee1f731 : SUCCESS in 7m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/527e6e0d04974aa79cd68ad0707d0a26 : SUCCESS in 6m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7d30a37a7644494d9dff19c65d927e71 : SUCCESS in 6m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7ef0bf1d84644725ba3e2070f58168f8 : SUCCESS in 1h 08m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/885a1391ab444aa9bbb787f8a06792b5 : SUCCESS in 51m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/41bd9b143c234906be66605525b2554c : SUCCESS in 44m 56s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b96b7bd994274fc8bf2ac4a131050102 : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9cd1cfe96c5a88aace061b3d9c4cdd51e7e3583b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-11 10:17:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14ff8fa12a6e38536b9005cca003379c959bb9ed","revisions":{"14ff8fa12a6e38536b9005cca003379c959bb9ed":{"kind":"REWORK","_number":1,"created":"2022-11-11 08:53:02.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/51/864251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/864251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/864251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/864251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/864251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/864251/1"}}},"commit":{"parents":[{"commit":"f89ffe6149f83fa43e924d5c3821579dba5d2419","subject":"Fix the state and default value when create cluster instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f89ffe6149f83fa43e924d5c3821579dba5d2419"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-11 08:51:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-11 08:52:09.000000000","tz":480},"subject":"feature: Add translation for monitor center","message":"feature: Add translation for monitor center\n\nAdded Chinese translation for monitor center\n\nChange-Id: Ib3f33ad45b5630b1b7d56573399873decb8a0a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/14ff8fa12a6e38536b9005cca003379c959bb9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/14ff8fa12a6e38536b9005cca003379c959bb9ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
