)]}'
{"id":"openstack%2Fskyline-console~864967","triplet_id":"openstack%2Fskyline-console~master~I61532a12312383cdedf2fdfca10633b16064f77b","project":"openstack/skyline-console","branch":"master","topic":"feat-magnum-redesign","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-18 06:29:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-18 05:54:10.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-11-18 09:38:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I61532a12312383cdedf2fdfca10633b16064f77b","subject":"feature: Support magnum in administrator platform","status":"MERGED","created":"2022-11-18 03:18:17.000000000","updated":"2022-11-18 09:39:41.000000000","submitted":"2022-11-18 09:38:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864967-feat-magnum-redesign","meta_rev_id":"48e4bdb9aa1f7283d45d6db7a487f0f016791775","_number":864967,"virtual_id_number":864967,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-18 09:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-11-18 06:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-11-18 06:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 03:21:22.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-18 03:21:22.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-18 05:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80cb73098bdfa0df4bd875dc471fced6ae9992d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-18 03:18:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c7ad98605560527098d002db4537e85b0b5c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-18 03:50:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4091d2aec359327bb2812f940b46dc3d250d625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 05:13:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b505c7bccb4c3ea5114a60157e89cf\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e26f9717824474aabf0427fcc55895 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbcf27275de4f3d863e95b25388a53f : SUCCESS in 5m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/bf401cfc5b4c464ba2afc0231a591b02 : SUCCESS in 7m 34s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e5c925aaafef4a31be18b7402afac17f : SUCCESS in 7m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ea3af1631923404fab26d9994356fff7 : SUCCESS in 6m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4fb4c26d51a04cb181cbfe4d8049251a : SUCCESS in 1h 20m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ac1232a90baf40438594abed9d7ea8c6 : SUCCESS in 1h 10m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9404e299e11e4e5499f6ee8d939370d9 : SUCCESS in 53m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c195db0b80724ad0b840e0222c6c363f : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0b2de6e60060193cb4fdc8d419c9c10de3379f48","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-18 05:54:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"72c115e0bce89c6272586a31a92a2ca4400e2ae4","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-18 06:10:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74b66c99b4e74a29deaaebd945551e4f948fbf93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-18 06:24:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8ccc55649d24601f57eb9a7566274e2c989aefa7","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-18 06:25:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"743b3d41431460ffa5c1c89f39f825087144e863","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-18 06:29:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"34f7d8826bb168bf744cba84950a58c53b06513a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 07:37:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411d6ac481bb4e4a93e014ad91b70560\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39bb53425b4c4923817e3ddf5c177e1b : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516fcb52278c44f498092d16a9b76c7a : SUCCESS in 4m 57s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fa71cf4a179d45f7810b5fbed7eea29b : SUCCESS in 6m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5f240eb601cb4a4bbc1d0083e40fc013 : SUCCESS in 5m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b293fff2cb11494faa078929e6745bae : SUCCESS in 4m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fae1e268c1db47b1bd6862f6e89ead28 : SUCCESS in 1h 10m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5767d4f4517f4730b79817ca23606181 : SUCCESS in 47m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ce0a54eb059a45eab41443775f000ae2 : SUCCESS in 41m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/051741dec42e46b2bf59de7069236ebc : SUCCESS in 48m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a2aab452e1c1c33f52bf1ed6cf2aa3ab93cc30df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 07:37:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f80a7fab4f3af197120ab675939f1e3431264e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 09:38:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7bf056db51d48e0a900199b53eb3eb2\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea9fee46e92645f3bb69565cab62d0f9 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493da808320a49138d1ebfab44fd42b0 : SUCCESS in 5m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/99ba1fbbf4264348a17e2d6b27013e6a : SUCCESS in 8m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7c69e82bd33a4888988d48d6e0db9620 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bdef57ce6410404e9dece9fc57fdeda2 : SUCCESS in 6m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4451b815d46a4c18b9eeb524f1a2800f : SUCCESS in 1h 54m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/050112fe607d4599850dcd5180023507 : SUCCESS in 54m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fad19e80a3c341d8a33457097fdccf39 : SUCCESS in 49m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/66d91c9e4af84942974b146ab36a3502 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":3},{"id":"98587b664220094118ce395719ad3cd902f06a6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 09:38:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"48e4bdb9aa1f7283d45d6db7a487f0f016791775","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 09:39:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2589d910f740efacf3af5d15d3e1e2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ef1744b38a7422aadd3b4247e97e908 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f831af6b9ef6412688171aed0c85b1df : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"47b3cead098702ec04feefa5bead3d9ef08f5fc6","revisions":{"ae09eaa8802c0400308f2d15105177a13e45472c":{"kind":"REWORK","_number":1,"created":"2022-11-18 03:18:17.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/67/864967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/864967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/1"}}},"commit":{"parents":[{"commit":"9df7b6ca0fac96a951380695e1a23ec9e283cb00","subject":"fix: fix the i18n of the status of vpn ipsec connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9df7b6ca0fac96a951380695e1a23ec9e283cb00"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 03:17:32.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 03:17:32.000000000","tz":480},"subject":"feature: Support magnum in administrator platform","message":"feature: Support magnum in administrator platform\n\n1. Show list and detail of cluster instance\n2. Show list and detail of cluster template\n3. Hide the keypair if in administrator platform\n\nChange-Id: I61532a12312383cdedf2fdfca10633b16064f77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae09eaa8802c0400308f2d15105177a13e45472c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae09eaa8802c0400308f2d15105177a13e45472c"}]},"branch":"refs/heads/master"},"a707c04e62dc065971821a4f0b4f455819833214":{"kind":"REWORK","_number":2,"created":"2022-11-18 03:50:07.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/67/864967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/864967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/2"}}},"commit":{"parents":[{"commit":"9df7b6ca0fac96a951380695e1a23ec9e283cb00","subject":"fix: fix the i18n of the status of vpn ipsec connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9df7b6ca0fac96a951380695e1a23ec9e283cb00"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 03:17:32.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 03:49:47.000000000","tz":480},"subject":"feature: Support magnum in administrator platform","message":"feature: Support magnum in administrator platform\n\n1. Show list and detail of cluster instance\n2. Show list and detail of cluster template\n3. Hide the keypair if in administrator platform\n\nChange-Id: I61532a12312383cdedf2fdfca10633b16064f77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a707c04e62dc065971821a4f0b4f455819833214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a707c04e62dc065971821a4f0b4f455819833214"}]},"branch":"refs/heads/master"},"47b3cead098702ec04feefa5bead3d9ef08f5fc6":{"kind":"REWORK","_number":3,"created":"2022-11-18 06:24:30.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/67/864967/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/67/864967/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/67/864967/3"}}},"commit":{"parents":[{"commit":"9df7b6ca0fac96a951380695e1a23ec9e283cb00","subject":"fix: fix the i18n of the status of vpn ipsec connection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9df7b6ca0fac96a951380695e1a23ec9e283cb00"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 03:17:32.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-18 06:24:13.000000000","tz":480},"subject":"feature: Support magnum in administrator platform","message":"feature: Support magnum in administrator platform\n\n1. Show list and detail of cluster instance\n2. Show list and detail of cluster template\n3. Hide the keypair if in administrator platform\n\nChange-Id: I61532a12312383cdedf2fdfca10633b16064f77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/47b3cead098702ec04feefa5bead3d9ef08f5fc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/47b3cead098702ec04feefa5bead3d9ef08f5fc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
