)]}'
{"id":"openstack%2Fskyline-console~865005","triplet_id":"openstack%2Fskyline-console~master~I4e340785ef37c06b45342ac38ccdce560414623a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-11-23 02:27:52.000000000","reason":"\u003cGERRIT_ACCOUNT_34589\u003e replied on the change","reason_account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"}}},"hashtags":[],"change_id":"I4e340785ef37c06b45342ac38ccdce560414623a","subject":"feat: add success and error status in Status component","status":"MERGED","created":"2022-11-18 09:56:26.000000000","updated":"2022-11-23 05:16:22.000000000","submitted":"2022-11-23 05:15:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865005","meta_rev_id":"6d0459a58044e49fb2cf4f44a95591a50f0dce53","_number":865005,"virtual_id_number":865005,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 05:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-21 02:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-11-21 02:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-18 11:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-21 02:12:15.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"a10219442f0f745ad2af7694de2823b9a40e1c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-18 09:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2fcea07092618717689c914d8f87a0e5560232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 11:20:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2afaf13a0f0472f911d1b97fd4c6d62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88de95f578a4616971eb5bdd1f29090 : SUCCESS in 5m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/913556abee26486b85754a7bf45eeaee : SUCCESS in 8m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4673b05bf77d415b8c3e7dd1bb869c1c : SUCCESS in 6m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ee7cdaee269149178309355fbe9b6951 : SUCCESS in 6m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ab07f8a923a44539adcb95d42f1bae8b : SUCCESS in 1h 21m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d43671a725064501b69d9d80660f4483 : SUCCESS in 49m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9cfdacedf5b246c78696fc79faacbbc0 : SUCCESS in 43m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b3d7869ea73544798b1783a6352e58a7 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3d6615c72b5e273c9a6581c4f6e255a37b5ee947","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-21 02:12:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9bead10d23e1ea8d70021f41cec63696d0fe0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 02:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"564a0744b5399bf0c56c4cf17b6abfe0df8bca85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 02:38:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c36dcee74d94431838aff1ea1775768\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ea54db4e664f4ca961d84149f82057 : SUCCESS in 6m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/48a925d8230b4a5fbd5b9891a8cc0ce4 : SUCCESS in 9m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f249c5b0383e443fafa871e3cc06cd65 : SUCCESS in 5m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/eabd3d5fc5cd42b28ada2d4a0c1fc3db : SUCCESS in 5m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/454d2904d9304866abd071a1510c65e0 : FAILURE in 19m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/19dc99f870194d05b57589e36d07b996 : FAILURE in 11m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/703568a483e8486d803f5c5c0baab046 : FAILURE in 15m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6f9bfdd0ac5f4d9981e4c32a422e7063 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8ff0c1ab89209e3d295b9896187b746fe70e835f","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-11-23 02:27:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac4abdfefa5fe3ba607d374751499b76e8e34fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 03:57:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50c2830f2c0c48d2aa1db29fbaefaf2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4e1ca47ffd401c991ed74c7002d159 : SUCCESS in 5m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cdb3f268e4924b338f2bee06a59f86e4 : SUCCESS in 10m 18s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9f547a32d713448d9e0499fb01afb9e3 : SUCCESS in 7m 31s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/423c0df5b776471c8356ccd305467928 : SUCCESS in 7m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/80360a803318486582c1101e7e3ed047 : SUCCESS in 1h 22m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fcc915257d9b437092195b7c02b1c60f : SUCCESS in 59m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ab98f5b8d685476d93b561b6ad964e72 : SUCCESS in 51m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/eaa47f829dfe4fa1823ac1f729348fa2 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ebe21d9ded00a2e4c1b082010a84cb61882c03db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 03:58:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dc756adfd3df316d02df37cc7a76084db283562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 05:15:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b4b6658fde4d9181e61bc97406d73c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76135df8a0234c68bbc32a91a1bf4e8a : SUCCESS in 5m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0f4831cc4ea34bb99b197fae3c8f9c78 : SUCCESS in 8m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a6995d0d92b74e5d8bacfded57b4ba45 : SUCCESS in 6m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/bb0d969a1adc4f1c81b1ca1d3b3631aa : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/21eb776277da445d96f48c7492ac39dd : SUCCESS in 1h 16m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf8f121b2d744876bd07920bddd804fd : SUCCESS in 54m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/76a8672245b649679650bd6ae0dd6f90 : SUCCESS in 46m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e1602157633a4767be33c00aff0139bf : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"07fc78c749ef022e0842cd533e34d26095112a3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 05:15:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d0459a58044e49fb2cf4f44a95591a50f0dce53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 05:16:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae976eaef544430a9c607f3e563bccd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30386da5fc834433bb2107b15ac57e99 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd30f8c308da5777497f3deb987610bc2743bab9","revisions":{"bd30f8c308da5777497f3deb987610bc2743bab9":{"kind":"REWORK","_number":1,"created":"2022-11-18 09:56:26.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/05/865005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/05/865005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/05/865005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/05/865005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/05/865005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/05/865005/1"}}},"commit":{"parents":[{"commit":"47b3cead098702ec04feefa5bead3d9ef08f5fc6","subject":"feature: Support magnum in administrator platform","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/47b3cead098702ec04feefa5bead3d9ef08f5fc6"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-18 09:50:49.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-11-18 09:50:49.000000000","tz":480},"subject":"feat: add success and error status in Status component","message":"feat: add success and error status in Status component\n\n1. add \u0027pass\u0027 key in successKeys\n2. add \u0027refuse\u0027 key in errorKeys\n\nChange-Id: I4e340785ef37c06b45342ac38ccdce560414623a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bd30f8c308da5777497f3deb987610bc2743bab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bd30f8c308da5777497f3deb987610bc2743bab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
