)]}'
{"id":"openstack%2Fskyline-console~865508","triplet_id":"openstack%2Fskyline-console~master~Ia0256668b0dbcd4dad31048b36929a99205707ed","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-image","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-24 10:09:17.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-24 10:08:20.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-11-24 11:15:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0256668b0dbcd4dad31048b36929a99205707ed","subject":"Fix: fix the image store in magnum node spec","status":"MERGED","created":"2022-11-24 07:40:24.000000000","updated":"2022-11-24 11:15:59.000000000","submitted":"2022-11-24 11:15:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865508-fix-magnum-image","meta_rev_id":"bec16a5c06e271f6ecc6a69a5649a3b318dd54fe","_number":865508,"virtual_id_number":865508,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-24 11:15:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-24 10:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-11-24 10:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-24 10:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 07:59:48.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-24 07:59:48.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-24 08:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e4496518cb21ad6f789c2782f3bdf57d8dc64ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-24 07:40:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7af30b941394e10db20219275bf3a042d8cc875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 08:46:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c7443a023f4130a40f3dc914b7af2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0aeb4767ac04b3a9f170ad2a82e903d : SUCCESS in 5m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8902fa993d5a486b995ed235521b91f1 : SUCCESS in 10m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/56edb0011b16419eaafd4ecfdc7baa68 : SUCCESS in 7m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e7f14f1c84c24b92bd6692ddaadf1d67 : SUCCESS in 7m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8a0c0734705949e3a711f3200d53ede0 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"874710e72eb136b608591912837bda88d136940e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-24 09:19:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d030c44d1714779d4ad11cf1c08e139c5abb581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 10:06:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b776dbc87964277a8dfea629396a076\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30132fad04294f7eb7dadf3376671792 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9047cb2c45744d6ebfc0ef042444a709 : SUCCESS in 8m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/98bab32e8d564d1ca268d3378262047c : SUCCESS in 6m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c2e9e1321e4242d3acb46253baf7dddf : SUCCESS in 6m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/680f64d5a3d24734adde8086b95ada61 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":2},{"id":"98aa81010f5e9e26ae76c899991a1eaeb144c754","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-24 10:08:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d4108811641a23a56bb25cb212fdcd61e3a230a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-24 10:09:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f7e610e6e3dc45758289c503815fa0fbce785453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 10:09:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36a6a5e0e25dbaac2d07c837e613cb914419fac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 11:15:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aea83b57da644638f8765abe3d767f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe665569d360437a9a5e57677017908c : SUCCESS in 4m 37s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1dc1bbfd8dde433eae4260cf11a0a490 : SUCCESS in 10m 33s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/32ded12e868b4ea98e35964dd193251d : SUCCESS in 7m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6dd12def5502487c97814ab4f2be8e19 : SUCCESS in 7m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/38173b69bfb54f32b4a7a97dff7a18a7 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7ec65f19c4ce094bd618e455bb479bfcb7471dcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 11:15:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bec16a5c06e271f6ecc6a69a5649a3b318dd54fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 11:15:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a88b22507b41e8b86d3ebce70fc75a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9469141b3ef4990b61598dec883e78b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72b81c645efe7dfaff17d23caba3187e7eb9dfa4","revisions":{"a26a0af9b88ab0dcf54e7a3d9e2009db82fa1677":{"kind":"REWORK","_number":1,"created":"2022-11-24 07:40:24.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/08/865508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/865508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/1"}}},"commit":{"parents":[{"commit":"463ced2667f4628dc45c04826e4fdc32e6b644e2","subject":"fix: Fix the store path in magnum and required name in zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/463ced2667f4628dc45c04826e4fdc32e6b644e2"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-24 07:39:59.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-24 07:39:59.000000000","tz":480},"subject":"Fix: fix the image store in magnum node spec","message":"Fix: fix the image store in magnum node spec\n\nfix the image store to get better extensions\n\nChange-Id: Ia0256668b0dbcd4dad31048b36929a99205707ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a26a0af9b88ab0dcf54e7a3d9e2009db82fa1677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a26a0af9b88ab0dcf54e7a3d9e2009db82fa1677"}]},"branch":"refs/heads/master"},"72b81c645efe7dfaff17d23caba3187e7eb9dfa4":{"kind":"REWORK","_number":2,"created":"2022-11-24 09:19:46.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/08/865508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/865508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/865508/2"}}},"commit":{"parents":[{"commit":"463ced2667f4628dc45c04826e4fdc32e6b644e2","subject":"fix: Fix the store path in magnum and required name in zun container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/463ced2667f4628dc45c04826e4fdc32e6b644e2"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-24 07:39:59.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-24 09:12:45.000000000","tz":480},"subject":"Fix: fix the image store in magnum node spec","message":"Fix: fix the image store in magnum node spec\n\nfix the image store to get better extensions\n\nChange-Id: Ia0256668b0dbcd4dad31048b36929a99205707ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/72b81c645efe7dfaff17d23caba3187e7eb9dfa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/72b81c645efe7dfaff17d23caba3187e7eb9dfa4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
