)]}'
{"id":"openstack%2Fskyline-console~865799","triplet_id":"openstack%2Fskyline-console~master~Ie0021cbddc4c8730d604e4892a8cfa878a5bcb95","project":"openstack/skyline-console","branch":"master","topic":"fix-create-keypair","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-11-28 10:14:37.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-11-28 10:14:11.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-11-28 12:11:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0021cbddc4c8730d604e4892a8cfa878a5bcb95","subject":"Fix keypair in magnum","status":"MERGED","created":"2022-11-28 08:01:08.000000000","updated":"2022-11-28 12:12:15.000000000","submitted":"2022-11-28 12:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865799-fix-create-keypair","meta_rev_id":"f24fd7b9180f3bf5089b018e2b29d5470c2f304a","_number":865799,"virtual_id_number":865799,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-28 12:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-11-28 10:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-11-28 10:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-11-28 10:14:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 08:12:03.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-11-28 08:12:03.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-11-28 10:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b32b3811256d986f41209eb8a8f1c991f5b7427d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-28 08:01:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eabe997f00e6b11c8e5f2ca1272faeeb012544a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-11-28 09:12:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"68315cdc403d7dd3ca4311804827a4498d53b45f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-11-28 10:14:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fc2bbb7d1ddb5731ddb11d7cd990cf5f8fd36a79","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-28 10:14:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"39e83094ba08416307a365266358b737f6a02574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:49:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5355e728e9ee4ca7832da894c71a3369\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1bab30cad64731bf4648a9585a2bbd : SUCCESS in 4m 49s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e40bc224dc3141e2adedad1512400ea4 : SUCCESS in 11m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8faf69c35510472d864eb7ce937c8d70 : SUCCESS in 6m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6accec66f6044f2597f612236502eb58 : SUCCESS in 8m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5bfb1688916b40438c6b640c76dce0e4 : SUCCESS in 1h 31m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e040242e9293405ba79dc0a77bde4861 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4fab6790f78991219e8c383b2993ceb0bad628da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 10:50:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e00eddc4e30a34e1338bd5038d06691375574beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:11:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2996d8ac20e497382c43f1072680585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7299408a509549a7aeac741be44a9ddc : SUCCESS in 6m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b07480ea98b8414ab6c54466008384b1 : SUCCESS in 8m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/da511e6921c14dcd8ecfc51153bf351d : SUCCESS in 6m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/69c6f62071d14ea996e2eed386a6ca28 : SUCCESS in 6m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3c03fc81dbbd41b7aa79d002a742b8ab : SUCCESS in 1h 19m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ce7c7bfdcd3c4ae4b2bbccf4b53e4500 : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0f9c1410c4d82f0675eb2d0194434d29066c918b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f24fd7b9180f3bf5089b018e2b29d5470c2f304a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 12:12:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cc9b3ea814c4cf7b46476fc81562b4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1fa777a8274ee586a431481f35bd4d : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac1edd6e9785c45b58a69debab99b9480a696177","revisions":{"7ff344af4e2e94d15fcfa27a7dcc5f712d9ee2eb":{"kind":"REWORK","_number":1,"created":"2022-11-28 08:01:08.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/99/865799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/865799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/1"}}},"commit":{"parents":[{"commit":"8401c5b5461f760b87e48844de1f0721f9163630","subject":"fix: fix the instance-volume in form items","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8401c5b5461f760b87e48844de1f0721f9163630"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-28 08:00:46.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-28 08:00:46.000000000","tz":480},"subject":"Fix keypair in magnum","message":"Fix keypair in magnum\n\n1.The form items will be resetted, when create a new keypair\n2.Same issue when create instance\n\nCloses-Bug: #1998077\nChange-Id: Ie0021cbddc4c8730d604e4892a8cfa878a5bcb95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ff344af4e2e94d15fcfa27a7dcc5f712d9ee2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ff344af4e2e94d15fcfa27a7dcc5f712d9ee2eb"}]},"branch":"refs/heads/master"},"ac1edd6e9785c45b58a69debab99b9480a696177":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-28 09:12:59.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/99/865799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/865799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/865799/2"}}},"commit":{"parents":[{"commit":"8401c5b5461f760b87e48844de1f0721f9163630","subject":"fix: fix the instance-volume in form items","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8401c5b5461f760b87e48844de1f0721f9163630"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-28 08:00:46.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-11-28 09:12:59.000000000","tz":0},"subject":"Fix keypair in magnum","message":"Fix keypair in magnum\n\n1.The form items value will be resetted, when create a new keypair\n2.Same issue when create instance\n\nCloses-Bug: #1998077\nChange-Id: Ie0021cbddc4c8730d604e4892a8cfa878a5bcb95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac1edd6e9785c45b58a69debab99b9480a696177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac1edd6e9785c45b58a69debab99b9480a696177"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
