)]}'
{"id":"openstack%2Fskyline-console~866547","triplet_id":"openstack%2Fskyline-console~master~I29bdfb3df6c859b2522b2a10a21173defce7765e","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-12-05 11:40:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29bdfb3df6c859b2522b2a10a21173defce7765e","subject":"fix: add middleComponent prop in KeyValueInput component","status":"MERGED","created":"2022-12-05 07:15:04.000000000","updated":"2022-12-05 11:40:58.000000000","submitted":"2022-12-05 11:40:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866547","meta_rev_id":"2e8e49ea539707487b602f5e8b5433c1b4654366","_number":866547,"virtual_id_number":866547,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-05 11:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-05 10:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-05 10:09:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-05 10:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-05 10:09:19.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"8cc23c0426544089488fb22adbd6c91cba12dd62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-05 07:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b07df3801cf67ed037abd4c2c2aeda048de9941a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-05 08:09:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ab841f559019878977f074a456eca0ff2e44dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-05 08:17:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bab2052b4af5a8d201f8738bc6ba6b968009da0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-05 08:32:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cd5c4670eff7dab5b9aeead4dc6dd7f35bba0361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-05 08:49:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ea70bee9ed76bb4bdaed3a7f7784bea0f450809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 10:04:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b8eff5c066a4db2a0e0b0ee6fa17a51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6af05a8186e489b80656843123d27d7 : SUCCESS in 6m 18s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6a1c5a2d8f1c45c888274021f76ffd94 : SUCCESS in 7m 29s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/764ba197ca424a5d93781092bc86c7aa : SUCCESS in 5m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/32e0a741205a43ffba53a476f7640cdd : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/31d214824b1a4069878f2b96fc9d8d3d : SUCCESS in 1h 13m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9a26fc5364c4476e8084c300f5ede2f0 : SUCCESS in 52m 57s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/090a2e2944d34ab785e3281691f3441f : SUCCESS in 46m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7d4d084e5d6043a080848c3df6084479 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":5},{"id":"3948d725a0d9f3eeb7eb4384e32f0f5710b7a575","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-05 10:09:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5288fdd5787d749c6efc42dfbbb523533aab1ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 10:09:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db88594c4711455b535a9fd08200211b634b89ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 11:40:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0628e530744d749088334411e52198\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4750e1359de447ac87789f436619ff91 : SUCCESS in 4m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/50e5248a5a1a4e1ab92d931a3ffd87e2 : SUCCESS in 10m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e82823c9704f44a4af91edac61dab6ac : SUCCESS in 6m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/42f5abc67e2b45deb8e65551961933f0 : SUCCESS in 7m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/aa9d0e20950c49468c247eb1dd9d8473 : SUCCESS in 1h 25m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/39a039681b2a41e89bbe78bf776fb026 : SUCCESS in 1h 06m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4ffcadcaf7e84a6aa3b0594c2e256090 : SUCCESS in 58m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dcc182c09af74892a31f6061a13d25a5 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":5},{"id":"6674d7a3fb9696a1fef77fe79feb1067b511a296","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 11:40:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2e8e49ea539707487b602f5e8b5433c1b4654366","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 11:40:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8bbc493029b48c59f0b0a0a9a9acdbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a200c717166415397c9aabfa6e31a86 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"719e26ebaa07f63edca2d73a7de6177b835de6b6","revisions":{"83091f23d3b7316f502d9c9b4e57d0e0fa5d64ff":{"kind":"REWORK","_number":1,"created":"2022-12-05 07:15:04.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/47/866547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/1"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"subject":"fix: add showEqualSign prop in KeyValueInput component","message":"fix: add showEqualSign prop in KeyValueInput component\n\n1. add showEqualSign prop in KeyValueInput component to support show or hide equal sign\n\nChange-Id: I29bdfb3df6c859b2522b2a10a21173defce7765e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/83091f23d3b7316f502d9c9b4e57d0e0fa5d64ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/83091f23d3b7316f502d9c9b4e57d0e0fa5d64ff"}]},"branch":"refs/heads/master"},"426c1940069bbe3dab062ba4dd27f93452ed66ad":{"kind":"REWORK","_number":2,"created":"2022-12-05 08:09:58.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/47/866547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/2"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 08:04:04.000000000","tz":480},"subject":"fix: add middleComponent prop in KeyValueInput component","message":"fix: add middleComponent prop in KeyValueInput component\n\n1. add middleComponent prop in KeyValueInput component to support show default equal sign or others\n\nChange-Id: I29bdfb3df6c859b2522b2a10a21173defce7765e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/426c1940069bbe3dab062ba4dd27f93452ed66ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/426c1940069bbe3dab062ba4dd27f93452ed66ad"}]},"branch":"refs/heads/master"},"b9b6aab075703d676206e6fc8929ead8a128cd3b":{"kind":"REWORK","_number":3,"created":"2022-12-05 08:17:30.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/47/866547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/3"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 08:17:16.000000000","tz":480},"subject":"fix: add middleComponent prop in KeyValueInput component","message":"fix: add middleComponent prop in KeyValueInput component\n\n1. add middleComponent prop in KeyValueInput component to support show default equal sign or others\n\nChange-Id: I29bdfb3df6c859b2522b2a10a21173defce7765e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b9b6aab075703d676206e6fc8929ead8a128cd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b9b6aab075703d676206e6fc8929ead8a128cd3b"}]},"branch":"refs/heads/master"},"65b091638b064683ef4941b56f29969d9e98a78b":{"kind":"REWORK","_number":4,"created":"2022-12-05 08:32:33.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/47/866547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/4"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 08:32:10.000000000","tz":480},"subject":"fix: add middleComponent prop in KeyValueInput component","message":"fix: add middleComponent prop in KeyValueInput component\n\n1. add middleComponent prop in KeyValueInput component to support show default equal sign or others\n\nChange-Id: I29bdfb3df6c859b2522b2a10a21173defce7765e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/65b091638b064683ef4941b56f29969d9e98a78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/65b091638b064683ef4941b56f29969d9e98a78b"}]},"branch":"refs/heads/master"},"719e26ebaa07f63edca2d73a7de6177b835de6b6":{"kind":"REWORK","_number":5,"created":"2022-12-05 08:49:06.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/47/866547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866547/5"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 07:13:24.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-05 08:48:49.000000000","tz":480},"subject":"fix: add middleComponent prop in KeyValueInput component","message":"fix: add middleComponent prop in KeyValueInput component\n\n1. add middleComponent prop in KeyValueInput component to support show default equal sign or others\n\nChange-Id: I29bdfb3df6c859b2522b2a10a21173defce7765e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/719e26ebaa07f63edca2d73a7de6177b835de6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/719e26ebaa07f63edca2d73a7de6177b835de6b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
