)]}'
{"id":"openstack%2Fskyline-console~866647","triplet_id":"openstack%2Fskyline-console~master~I90d1bc3c2bcb2aa95d728c245dd1b742d3fe6ae1","project":"openstack/skyline-console","branch":"master","topic":"feat-magnum-quota","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-12-06 09:49:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-06 09:48:18.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-06 10:57:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90d1bc3c2bcb2aa95d728c245dd1b742d3fe6ae1","subject":"feature: Support quota in magnum service","status":"MERGED","created":"2022-12-06 06:15:41.000000000","updated":"2022-12-06 10:58:50.000000000","submitted":"2022-12-06 10:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866647-feat-magnum-quota","meta_rev_id":"6f9f361526e0d912f6ec743fc5873082a9d5a73a","_number":866647,"virtual_id_number":866647,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-06 10:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-06 09:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-12-06 09:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-12-06 09:49:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 07:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 07:49:41.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-12-06 07:49:41.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"f4953407ad4e15629e6808c0da421e6b38903e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-06 06:15:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bbf8168b1ae1f2e61f2f7df0839b6b1f5fb8f70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-06 06:24:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f3dbf30bfc77cab22e3e4ea8d1854cf76916417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 07:49:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9199440e7e9f446887174a41cf20f935\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73bcc8c45a354baf8fde0bb14a9999ff : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ce37c0fa074640956acc3af345e71e : SUCCESS in 5m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/130502b1c1c048898ec5dc0df1cb756f : SUCCESS in 10m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b0ca3c0c46bf4558beaee85a842557da : SUCCESS in 6m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/667a535f9a074cd3bc672f8d93705b0e : SUCCESS in 7m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/70a3a3ae920642ecbdec71abd99229b2 : SUCCESS in 1h 23m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c01384357fd347829da571a484b68ed5 : SUCCESS in 1h 08m 13s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3ffa7618b60c4e18a88edda43aeaa5a5 : SUCCESS in 56m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3b40fd5edb104b4da8956d900446c603 : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":2},{"id":"af1408aac7c5ecbffb13db1994808036ffdffdba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-06 08:08:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"187f6638e3f5ce18abe88c8cd58569c754ffe696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 09:46:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126a27b3ec324d60ab879898da1ecbb3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e927a2ae1024c7f9900b344f99f3479 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db5fec4b42a847c5938607720117a908 : SUCCESS in 4m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4cffed0c99494297909b16685998a0ca : SUCCESS in 9m 59s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3cfdb032b14040998b07139b0a63a6c3 : SUCCESS in 7m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b365f711a4554d0fae18a0e47d9971de : SUCCESS in 7m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/017292734b3448329b7f85a9e1947238 : SUCCESS in 1h 32m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dfee174d12fb4412bce140203d8b5d9f : SUCCESS in 55m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/06509f57246a4103a7ed6076a8ee93d1 : SUCCESS in 49m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2400e02e5523437e89fc09d877a5516a : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9b2491e78fc3894bf6915fd5f2138cfc36623923","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-06 09:48:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"21a17e4223c836d0b0ff0d9918aded2bfaa7af99","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-06 09:49:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd152e5568e2ece6bd2dbd50ac510a4cc97dba81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 09:50:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0fb9788d0838add423105ce289b690a4c4506675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 10:57:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17a63637729e47cb845963cb3222643a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c851fa48fb9844f4a1dfd333a91b08dd : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843e36ec95f84fdb9abde554c0825fb8 : SUCCESS in 5m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8808523d9e06420da0c05d13d0e4709f : SUCCESS in 7m 08s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e8548d83cdac4bde882d8ee227103f20 : SUCCESS in 6m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5a84718d659f43aca2b91f77fa7f8b4c : SUCCESS in 6m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/18de79a2eae84d8da81097f223f7cdda : SUCCESS in 1h 05m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0b7fe1f10175460bab76f2c25ca8496e : SUCCESS in 51m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2265240bea0f43439c2170252084adce : SUCCESS in 50m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e9cdd53722dc4faaac08368fac7c6e16 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":3},{"id":"390fc2b359a2ced88a6765d55fa4d7c77964b6a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 10:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6f9f361526e0d912f6ec743fc5873082a9d5a73a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 10:58:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d17d5d52eae429e867c99eefbfe0091\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5fb86a3313c4c30817d56a9db910e32 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4a6d572db842a59cd9cee7fea1b480 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a4d17ee190b7d44e8b2ee51d5423569f405739b","revisions":{"59fe17141aef0d1eb71bb38cfc185383b192e60a":{"kind":"REWORK","_number":1,"created":"2022-12-06 06:15:41.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/47/866647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/1"}}},"commit":{"parents":[{"commit":"b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0","subject":"fix: fix health_status in magnum cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4f23fd78c6bc4223e9ee8e7c2f54a660a0200b0"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-05 07:07:20.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-06 06:14:28.000000000","tz":480},"subject":"feature: Support quota in magnum service","message":"feature: Support quota in magnum service\n\n1.Show quota info when create cluster\n2.Support quota setting of magnum cluster when manage project quota\n\nChange-Id: I90d1bc3c2bcb2aa95d728c245dd1b742d3fe6ae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/59fe17141aef0d1eb71bb38cfc185383b192e60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/59fe17141aef0d1eb71bb38cfc185383b192e60a"}]},"branch":"refs/heads/master"},"b1f3c2066eecb5dfd4fe40ff766c987ef30af344":{"kind":"REWORK","_number":2,"created":"2022-12-06 06:24:39.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/47/866647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/2"}}},"commit":{"parents":[{"commit":"719e26ebaa07f63edca2d73a7de6177b835de6b6","subject":"fix: add middleComponent prop in KeyValueInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/719e26ebaa07f63edca2d73a7de6177b835de6b6"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-05 07:07:20.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-06 06:23:49.000000000","tz":480},"subject":"feature: Support quota in magnum service","message":"feature: Support quota in magnum service\n\n1.Show quota info when create cluster\n2.Support quota setting of magnum cluster when manage project quota\n\nChange-Id: I90d1bc3c2bcb2aa95d728c245dd1b742d3fe6ae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b1f3c2066eecb5dfd4fe40ff766c987ef30af344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b1f3c2066eecb5dfd4fe40ff766c987ef30af344"}]},"branch":"refs/heads/master"},"7a4d17ee190b7d44e8b2ee51d5423569f405739b":{"kind":"REWORK","_number":3,"created":"2022-12-06 08:08:05.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/47/866647/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/47/866647/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/47/866647/3"}}},"commit":{"parents":[{"commit":"719e26ebaa07f63edca2d73a7de6177b835de6b6","subject":"fix: add middleComponent prop in KeyValueInput component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/719e26ebaa07f63edca2d73a7de6177b835de6b6"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-05 07:07:20.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-06 08:07:48.000000000","tz":480},"subject":"feature: Support quota in magnum service","message":"feature: Support quota in magnum service\n\n1.Show quota info when create cluster\n2.Support quota setting of magnum cluster when manage project quota\n\nChange-Id: I90d1bc3c2bcb2aa95d728c245dd1b742d3fe6ae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a4d17ee190b7d44e8b2ee51d5423569f405739b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a4d17ee190b7d44e8b2ee51d5423569f405739b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
