)]}'
{"id":"openstack%2Fskyline-console~866829","triplet_id":"openstack%2Fskyline-console~master~Ic6faebfa21b92479b982770a00046a54882f21d8","project":"openstack/skyline-console","branch":"master","topic":"feat-magnum-quota","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-12-07 09:35:28.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-07 08:51:51.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-07 09:35:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6faebfa21b92479b982770a00046a54882f21d8","subject":"fix: Improve quota in magnum service","status":"MERGED","created":"2022-12-07 07:56:28.000000000","updated":"2022-12-07 09:36:33.000000000","submitted":"2022-12-07 09:35:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866829-feat-magnum-quota","meta_rev_id":"340d15772d2919c65316370b0b1e9ffbb0883c24","_number":866829,"virtual_id_number":866829,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 09:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-07 08:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-07 08:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 08:00:06.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-12-07 08:00:06.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-12-07 08:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e3ec3dc3dcf6265351b9ed99a457e0d83e2be0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-07 07:56:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03e0ac1a9e1b6a808506d26cddf3a4c4aec66cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-07 07:57:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc8ba8ea908c44f2b8c855e89b077f7c9506e4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 08:45:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ddc69d830ab4621ac2a1c38b5e8f49a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856c1c800b1d48249fb42a7eda1dff30 : SUCCESS in 5m 32s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/09a5778030644159ac4ef7d8c3ae45e5 : SUCCESS in 10m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b7a420735a5c4f689d6fd368ab8a6881 : SUCCESS in 6m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9057b668e1094fb7b8d7e43ce3a9b56d : SUCCESS in 5m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/25a9744c59b0484c83013b720b5e8b23 : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":2},{"id":"36de4b4e4d9c9f65ca7d3a8f598e49e8afa16f50","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-07 08:51:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c1cd2b3c1942a6c2e26c64294be940a988e16094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 08:52:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4019942fadff49ef44fba80965d3fbc06b4d853c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:35:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7ddf3a897224e299a68ca74d708d55d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614d5b6a37b54f7a802974b4e698ef3a : SUCCESS in 5m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/186e154390e84ef6bd0a0964e5bd63b1 : SUCCESS in 11m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6209ff234a3d4476abbc260a9847ea20 : SUCCESS in 7m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e87efbaeac3748889f88d920b4571788 : SUCCESS in 5m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/94e808c98f6d4c51af52f0a1f6667f4c : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d934f3c312f36ecba19e08efa70a2adc91fd88ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:35:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"340d15772d2919c65316370b0b1e9ffbb0883c24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:36:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87ee8a642d3a4d7d96c45d7f38d857a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d3fa48ccbe42b695a75e0b61dca68a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"228ccff07b3e760ed835a8e85cc62e8c7fd99f18","revisions":{"81e72835e720379a73b0193f9f155e73f36ae3ba":{"kind":"REWORK","_number":1,"created":"2022-12-07 07:56:28.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/29/866829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/866829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/1"}}},"commit":{"parents":[{"commit":"a2d170e2850c95f84013d6bec6b42349d848dbc0","subject":"Merge \"feat: support component in the table header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2d170e2850c95f84013d6bec6b42349d848dbc0"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-07 07:55:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-07 07:56:11.000000000","tz":480},"subject":"feature: Improve quota in magnum service","message":"feature: Improve quota in magnum service\n\n1.Show quota info by template when create cluster if did not select flavor or master flavor\n\n2.Remove link of keypair if the keypair is not belong to the current user\n\nChange-Id: Ic6faebfa21b92479b982770a00046a54882f21d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81e72835e720379a73b0193f9f155e73f36ae3ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81e72835e720379a73b0193f9f155e73f36ae3ba"}]},"branch":"refs/heads/master"},"228ccff07b3e760ed835a8e85cc62e8c7fd99f18":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-07 07:57:35.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/29/866829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/29/866829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/29/866829/2"}}},"commit":{"parents":[{"commit":"a2d170e2850c95f84013d6bec6b42349d848dbc0","subject":"Merge \"feat: support component in the table header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2d170e2850c95f84013d6bec6b42349d848dbc0"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-07 07:55:44.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-07 07:57:15.000000000","tz":480},"subject":"fix: Improve quota in magnum service","message":"fix: Improve quota in magnum service\n\n1.Show quota info by template when create cluster if did not select flavor or master flavor\n\n2.Remove link of keypair if the keypair is not belong to the current user\n\nChange-Id: Ic6faebfa21b92479b982770a00046a54882f21d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/228ccff07b3e760ed835a8e85cc62e8c7fd99f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/228ccff07b3e760ed835a8e85cc62e8c7fd99f18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
