)]}'
{"id":"openstack%2Fskyline-console~867352","triplet_id":"openstack%2Fskyline-console~master~I10dc6ebc3d55305ac4829677c0edd20a549a8662","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"34589":{"account":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"last_update":"2022-12-13 13:19:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10dc6ebc3d55305ac4829677c0edd20a549a8662","subject":"fix: add disabledRemoveFunc prop in AddSelect component","status":"MERGED","created":"2022-12-13 10:45:28.000000000","updated":"2022-12-13 13:25:48.000000000","submitted":"2022-12-13 13:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867352","meta_rev_id":"7ef87bd09ed99c7a0dc86d303cc41ff936aeac93","_number":867352,"virtual_id_number":867352,"owner":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 13:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-13 10:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-13 10:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-13 10:58:45.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-12-13 12:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9173124f3652859c3d0437239072394b36294f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"date":"2022-12-13 10:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dd38c3bb00943e8fd4376c7bbe7a10c86d274a7","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-13 10:58:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a76b19f3b076cbe21cbcb4c8969001e34821a343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 12:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e515b4b6044b53949be49b52c00aff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3a04a98efe4780aa9a638a4e954280 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9b8151756ea647818d2edd2716e35c8a : SUCCESS in 8m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ffba1149211b43a3b9466400529e3ed9 : SUCCESS in 5m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/76a0a76d32424cb9bf5b6f6206a6dc35 : SUCCESS in 6m 21s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6526c0779bf747658b1a52e224afa939 : SUCCESS in 1h 06m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7ba12c7e5a9d41bb9241c4e57dea62fe : SUCCESS in 1h 11m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f74a872671914d658af72a5a75f1cd70 : SUCCESS in 43m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/11e0101b4db544b69e21466122e87164 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aadd853f4b629d552314e466b63a5d519a3b6dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 12:03:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00c608a48855b938bfea163195d5165391d06108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:19:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72c801f31e6469c97cb97a7497688e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9382373de7b4335afb4c7b892639aaa : SUCCESS in 4m 57s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/46b375c01bb04d00b34b3c8cdd5c0b6a : SUCCESS in 9m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ead5751f8ec44ba29b41a87ab3b872b9 : SUCCESS in 6m 55s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1ec89f01719e4be3b1402f2fe35e7ec5 : SUCCESS in 7m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/214062e410f24ef780825617572a9c05 : SUCCESS in 1h 14m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8a5664f915824908828d942c2b0a2db8 : SUCCESS in 58m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e5c5e43e08e24f71b60ed8b1cff3d3e7 : SUCCESS in 45m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/87ebe817aff046d1aac0734a143af1bd : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ca6bf45979a3a7d00f5f6777e1be7c4f1d702e4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ef87bd09ed99c7a0dc86d303cc41ff936aeac93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 13:25:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bdb804ddf814476a0b593ffeb7c6bee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e37714839604493874e71263e1de41d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"124fdc3c2aaee3fa82eb4c8ba63c85a4136a2630","revisions":{"124fdc3c2aaee3fa82eb4c8ba63c85a4136a2630":{"kind":"REWORK","_number":1,"created":"2022-12-13 10:45:28.000000000","uploader":{"_account_id":34589,"name":"zhangke","email":"zhang.ke@99cloud.net","username":"zhangke"},"ref":"refs/changes/52/867352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/52/867352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/867352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/867352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/867352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/52/867352/1"}}},"commit":{"parents":[{"commit":"3ada93aae8cf6c21d6e801143473247acf0ca537","subject":"fix the validator of cluster name and labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ada93aae8cf6c21d6e801143473247acf0ca537"}]}],"author":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-13 10:43:55.000000000","tz":480},"committer":{"name":"zhangke","email":"zhang.ke@99cloud.net","date":"2022-12-13 10:43:55.000000000","tz":480},"subject":"fix: add disabledRemoveFunc prop in AddSelect component","message":"fix: add disabledRemoveFunc prop in AddSelect component\n\nAdd disabledRemoveFunc prop in AddSelect component to handle if disable the remove button and edit item\n\nChange-Id: I10dc6ebc3d55305ac4829677c0edd20a549a8662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/124fdc3c2aaee3fa82eb4c8ba63c85a4136a2630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/124fdc3c2aaee3fa82eb4c8ba63c85a4136a2630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
