)]}'
{"id":"openstack%2Fskyline-console~867783","triplet_id":"openstack%2Fskyline-console~master~I586689d508cb838562ba4a0bef574a4e696e57c0","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-16 02:50:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I586689d508cb838562ba4a0bef574a4e696e57c0","subject":"fix: fix the magnum service","status":"MERGED","created":"2022-12-15 11:00:47.000000000","updated":"2022-12-16 02:51:33.000000000","submitted":"2022-12-16 02:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867783-fix-magnum-bugs","meta_rev_id":"cc26ceed10c67aa8aa3c189e2b883598711de7f7","_number":867783,"virtual_id_number":867783,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 02:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-16 01:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-16 01:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 12:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 01:39:00.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"aa255290097f9c217fa4ec10559f700dc65e9806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-15 11:00:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2daa74e0d048efb470e463fe8b953c2286951969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 12:12:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e4bcaee3a04104a8695929054e5e84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9194cabc177446babd7e60644a010ca1 : SUCCESS in 4m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/84c6cb92ea6b4e2090ea10c453e8c99b : SUCCESS in 9m 43s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6ef1226e8b0f4d0bad33412de923c846 : SUCCESS in 6m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4c01ac5ad5914eeb91bad32c452bbc06 : SUCCESS in 7m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/22b9bb55f9d44d63ac0bb8afbf658d70 : SUCCESS in 1h 04m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/215ce4e885ac454d8f2a6fd3c4acfa24 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b310e63092c0a5df19fafb7f6e8586ecb16b0cdb","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-16 01:39:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29cb12472e5572d6cd9a5b4b20e5df25b490ea8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 01:39:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d7df308c2b96165c5a751455f5fce7f300c6028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 02:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448df37b15534d94b3b392a7362c9af9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b709feef3f31445c95cc331cb53d5b1d : SUCCESS in 5m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0c8aef800f1f4436a173f5ae925cd393 : SUCCESS in 10m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ce209974130945c889cdc4695aa8dc82 : SUCCESS in 7m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ba595d27cd62432b86b7001478c52fad : SUCCESS in 7m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/017e71aa36aa42f0aacfa7e96f5b2957 : SUCCESS in 1h 05m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/dfba6496308349baaee407e4f1f83a9a : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1013d058bdcdf6931afcac9d1020f92185abca58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 02:50:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc26ceed10c67aa8aa3c189e2b883598711de7f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 02:51:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa08bc87f10743758d9a316611140e36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41960ad9d3a42e2a8323fed351449e6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a307f457e4d90f58789bf6d5ef1ba7b85ddd0284","revisions":{"a307f457e4d90f58789bf6d5ef1ba7b85ddd0284":{"kind":"REWORK","_number":1,"created":"2022-12-15 11:00:47.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/83/867783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/83/867783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/867783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/867783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/83/867783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/83/867783/1"}}},"commit":{"parents":[{"commit":"f384bd4b73fde41f775104b84fe2bf58337827fa","subject":"feature: Add listener input event for instance-volume component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f384bd4b73fde41f775104b84fe2bf58337827fa"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-15 10:23:57.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-15 10:59:28.000000000","tz":480},"subject":"fix: fix the magnum service","message":"fix: fix the magnum service\n\n1. check cluster quota immediately when create page mounted\n2. fix template coverage to instance information\n3. fix detail fetch to show image name,flavor name...\n\nCloses-Bug: #1999756\nChange-Id: I586689d508cb838562ba4a0bef574a4e696e57c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a307f457e4d90f58789bf6d5ef1ba7b85ddd0284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a307f457e4d90f58789bf6d5ef1ba7b85ddd0284"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
