)]}'
{"id":"openstack%2Fskyline-console~868051","triplet_id":"openstack%2Fskyline-console~master~I3c2120593ff533d38fdc7cb026d0d2d3bf1e7132","project":"openstack/skyline-console","branch":"master","topic":"yaml","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-01-06 10:03:54.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I3c2120593ff533d38fdc7cb026d0d2d3bf1e7132","subject":"fix: fix the getValue of yaml","status":"MERGED","created":"2022-12-19 03:46:24.000000000","updated":"2023-01-06 12:55:03.000000000","submitted":"2023-01-06 12:54:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868051-yaml","meta_rev_id":"48abda262ebfea8f084b3055a97662f3a33840a1","_number":868051,"virtual_id_number":868051,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-06 12:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2022-12-19 05:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2022-12-19 05:25:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2022-12-19 05:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 05:21:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 05:25:17.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2022-12-19 05:28:40.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"c1ca67d130191655578008e2525cc4ce403fc2d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-19 03:46:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"709a29a4e8eb9ab21cf9b61176162c482cffea24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 05:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb925f13ce2d4715836ead9fc57d6763\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6842583e914179bd69cfd9030780df : SUCCESS in 5m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7380596db60348abbdef47feec100029 : SUCCESS in 10m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dde96b2ddd8b4fc9b23a3a2384ef2221 : SUCCESS in 7m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ca23b4801112453a98142c8c190e6e2a : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f06cb6936d17406b85d5117b61f3a193 : SUCCESS in 1h 27m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/26e773ef572c4e6ea4d0dc1b50d8e39f : SUCCESS in 1h 02m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1f59ff106ed64fea806f5b469f950878 : SUCCESS in 1h 01m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0209a50fed4c486182e8a9baae293eb3 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"66c199f6daf45ea75b3d27a3fc864ea0b5917344","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-19 05:25:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef60ce5d7cdc0246df72ffab422c6a1404014082","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-19 05:28:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8206008ae35d46ed6c55e329b3c9da3e6a3889fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 05:29:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb16b8683b333c8700ff0459a218285962d47905","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 07:02:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75b22115dffd406182435c0cb45af531\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57dbeb7d298462facf13a7a4405ed7f : SUCCESS in 5m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8a31267cecda4d6aa71e1b13ae7eae70 : SUCCESS in 6m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8467dd3a37c14a29bff99cc52b70a699 : SUCCESS in 7m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/82c4d1b5ec04433fbd11cdd28df2433c : SUCCESS in 7m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ebca1bbbb7964f29a1a0e11c03ceb412 : SUCCESS in 1h 28m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5d169cae1076447493e2e264f2b05fa8 : FAILURE in 45m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ed2d3501376c45459aaf8b599f1ed891 : SUCCESS in 52m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3eb381977f054499b7cf7c75bdbe8d85 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8af44ca28c708c0b634fd78a37b41f74c5a35980","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-19 07:35:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c607636f895bd06836b8f2814930e01167302c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:01:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d4af6faa503426abbbf45e1b4a3bd73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7549536b18ac4321927927370fb1faf2 : SUCCESS in 5m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/45cadfc3e3bf42d7a1fb51cc317c0b08 : SUCCESS in 8m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/44f09719bed94c3d9d649cdad413bff5 : SUCCESS in 5m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5da67328e6c9484d82ed3d393a5e2800 : SUCCESS in 6m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/44266e49ce9b46f38c26955c9a897369 : SUCCESS in 1h 19m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a65b67e241d84a829582e00cfe35635a : FAILURE in 1h 20m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/35f82fcd711645d8913e98df7e49678d : SUCCESS in 1h 03m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5cd29a2c40204c23942ee0fa28c1a381 : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"30af78d8e0d7cb6758cc02fcc6e6349699fdcb6b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-20 01:38:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8442b57e1e21d7ebf2636204749df98badc5da2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 02:51:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91df6e19a9b42f3895af823f42f8ba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97224e3b0c83471eb38c859eef7d09bf : SUCCESS in 4m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e619a802acc4408583c775946ec1ae50 : SUCCESS in 6m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/70c7f6913b5045e9b1d1bc5979c1722f : SUCCESS in 4m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f74ec576905c4e7683cf11cccb4a2ddf : SUCCESS in 5m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/78e096354a0a4021a4df3539b244c93c : SUCCESS in 1h 11m 50s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/562e04ba976042ae8f6f7b5f6a9c5c10 : SUCCESS in 45m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1ea38e705c514058b669066523647159 : SUCCESS in 40m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/16e4f77f0230474a93e3bd4a85128ac5 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":1},{"id":"41a3283436948733319ffc34dd81cacadd72a208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 02:52:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df7c669766344b9590bcf79ffb38933ca85c44d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 04:04:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f507f0d8d93f4d1686bef7e841972a09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91c86cea248447cb245dfbee2486f92 : SUCCESS in 5m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e0cc56adb5cb46ac83b16f12f3e78740 : SUCCESS in 11m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/109e940797eb458986642ec56c1c0a13 : SUCCESS in 5m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/eefbec30badc4c6db05cb86400af7c23 : SUCCESS in 6m 18s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/06753684a8534e7dbc6bb73674ac5098 : SUCCESS in 1h 09m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3c6b9d7c081448a4ad0ad9ba7095d78f : FAILURE in 51m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/203f0ea4caaa4ef2beb8f4a73e08fe7a : SUCCESS in 47m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/291ae3f2f880413ea62ca469ee1aac5f : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"44220490573fb76147f61bc8785ca5ed6c32e186","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-20 07:18:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d1c1f8c26457d7ec34d1257ab61ff7a5efa3777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 08:53:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c2db8e5bbc436ab31ae024ae160525\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547597a94404435fb9553c8e338d781c : SUCCESS in 5m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/59040914f50a4cf5996842e54d5d0ab7 : SUCCESS in 6m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6adf3ef1b0d742e39866b5fd346d1170 : SUCCESS in 7m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9eec9ad297a340d58b9bdb3a89c891bc : SUCCESS in 4m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/db0344d7ef90475b8e42bd699bf7b746 : SUCCESS in 1h 28m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/06e86001810340fbbd808b6012ec120b : FAILURE in 1h 08m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d0ad68b41ff54220bf362854a6790f48 : SUCCESS in 43m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/33c99b7be7ce49ba89b66591fc45b4dd : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"474d6bbd1c08f671402b8fc81a549d30cbf365e6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-01-06 10:03:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0115b754865b352c57656aaf94b489a2d0345a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 11:31:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e929e9fa783462a85166eb8440abcbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35bcd5befc824fa78d59198c8b14c5e2 : SUCCESS in 3m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4c7f3084926a48bba4c0b30355c1c401 : SUCCESS in 9m 33s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a86a4f16f66043868a2d84929d7d9aec : SUCCESS in 6m 45s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d9235482760e44718aec924e826c47f0 : SUCCESS in 6m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6cfd1fc662604d2da5a48bf50fc7ac3c : SUCCESS in 1h 22m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7bef8ded08b84a2cb3ba198c6a7829fa : SUCCESS in 1h 01m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/856eca5c0e9f49578222f7b6446437b0 : SUCCESS in 51m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/41f26ab5c4b54b9e8d518a9550f1087b : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9bf79fccb20b466eed575d455ab376767cb8b594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 11:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e288f9d8ae0c40fcdd5f9c2005ab18cc750051d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 12:54:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc342e8064654aa689c84a45dc7ef583\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3051600b9065493393089fb0a91f1f01 : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/52577017fe334a45b95fccd924c75d4d : SUCCESS in 8m 22s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/41d6456978a3473fad73ccb0219b97f1 : SUCCESS in 5m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e8750d1ed2824211b6bb0d8c1c1a5122 : SUCCESS in 5m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ac602f42e60a4bfaa8465376a8b34546 : SUCCESS in 1h 21m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3596c3f6c56f4c99a03326980fed1af9 : SUCCESS in 1h 04m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/89b2492912e542dbbfb177c1ecbb8c80 : SUCCESS in 57m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a015b66c006c43d089c8caafd6373285 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"83f5080f8810999e108cb3bb6181e43b502a6180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 12:54:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"48abda262ebfea8f084b3055a97662f3a33840a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 12:55:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69058f4a48d14480b72adefff3c543d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227aa88d866a4b39bcffbeb46c0f0e10 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4a24d04adaa65bf67c5bce2ab7fcec0bb72c33d","revisions":{"f4a24d04adaa65bf67c5bce2ab7fcec0bb72c33d":{"kind":"REWORK","_number":1,"created":"2022-12-19 03:46:24.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/51/868051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/868051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/868051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/868051/1"}}},"commit":{"parents":[{"commit":"a307f457e4d90f58789bf6d5ef1ba7b85ddd0284","subject":"fix: fix the magnum service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a307f457e4d90f58789bf6d5ef1ba7b85ddd0284"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-12-19 03:46:02.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2022-12-19 03:46:02.000000000","tz":480},"subject":"fix: fix the getValue of yaml","message":"fix: fix the getValue of yaml\n\n1. Fix the yaml value dump in the utils\n2. Update the unit test\n\nChange-Id: I3c2120593ff533d38fdc7cb026d0d2d3bf1e7132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a24d04adaa65bf67c5bce2ab7fcec0bb72c33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f4a24d04adaa65bf67c5bce2ab7fcec0bb72c33d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[]}
