)]}'
{"id":"openstack%2Fskyline-console~868056","triplet_id":"openstack%2Fskyline-console~master~I18457d3593761e36a7173488b646ebdedf656ec2","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-19 08:12:05.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-19 09:53:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18457d3593761e36a7173488b646ebdedf656ec2","subject":"fix: fix the delete operation of cluster instance","status":"MERGED","created":"2022-12-19 06:59:54.000000000","updated":"2022-12-19 09:54:57.000000000","submitted":"2022-12-19 09:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868056-fix-magnum-bugs","meta_rev_id":"036a4900c604529c42cb37d259212e756fcd1abe","_number":868056,"virtual_id_number":868056,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 09:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-19 08:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-19 08:12:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 08:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 08:03:54.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"ea2cf285e83ba14de00d17cae4cb69e76bb06959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-19 06:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae40c55073a033797e7c87d602b21b9975bad36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 08:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6dd37b0acb4462b351ba3b1d7b0742\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0432f5607c45319618ebc02613d724 : SUCCESS in 4m 57s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/473e540fe75b4b3caf86385d6ab72570 : SUCCESS in 9m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/eb5b112098d84c7f894f18a34a37ba4f : SUCCESS in 6m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a4531b9f60dc40039579d7c82489a731 : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d5bde606c0694ddda46892cd94b460ff : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8da5cacf5b9b57edf227f4db55018747d6022d0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-19 08:07:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"190a5ffa41d6dbe8f64244b622420ad66b8f1382","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-19 08:12:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83347fceba61947ed522c93e2309a2825ef092a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 08:47:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3f7089b7fd943288232b23aa7fa2cb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a156531e6b42b5b6cf0ae0330b272e : SUCCESS in 6m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b578013f25924911b64d12cf6e0f4181 : SUCCESS in 8m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7857b157d9d648498e8333b5180b90e7 : SUCCESS in 6m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2219791eb6d54d748298eca12051d311 : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7d63034bf77546618bb808109012d47f : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0ceae133ddaf889d30fba72132e80f98b22eebd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 08:47:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a666dd1ee903064180b3c7338526fbb6976046a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:53:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/105c76a36a36413ba14268abb6e7fb30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80d33897a0e4f60af46477cc41274bf : SUCCESS in 5m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e7f754cf3fc045b280b96f8e0074a8f1 : SUCCESS in 10m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e2d62b29ac894970afe52d643fb9db09 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4782bf962c43474491ccbc0211328769 : SUCCESS in 7m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4d4b6975131040ccb5600b8b3e0bec9c : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5d3c00940b7200bf6d75ae7f9f4b3382007803de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:53:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"036a4900c604529c42cb37d259212e756fcd1abe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 09:54:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e97ea18840a43c3802c906475cff763\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a65889df4244d29b1662f10c172e58 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b6e8253a6237e3f5dfc7506fb222ee83a47f68b","revisions":{"3ce471d81d705efa6f2f42cf75cadbb883922f76":{"kind":"REWORK","_number":1,"created":"2022-12-19 06:59:54.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/56/868056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/868056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/1"}}},"commit":{"parents":[{"commit":"2d437e5b4e6e2b1a2a40ef86105c14d43a548883","subject":"feat: support custom render in the TabDetail component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d437e5b4e6e2b1a2a40ef86105c14d43a548883"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-19 06:59:12.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-19 06:59:37.000000000","tz":480},"subject":"fix: fix the delete operation of cluster instance","message":"fix: fix the delete operation of cluster instance\n\nDisable the delete operation if the stack is not ready or the status is `DELETE_IN_PROGRESS`\n\nChange-Id: I18457d3593761e36a7173488b646ebdedf656ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ce471d81d705efa6f2f42cf75cadbb883922f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3ce471d81d705efa6f2f42cf75cadbb883922f76"}]},"branch":"refs/heads/master"},"7b6e8253a6237e3f5dfc7506fb222ee83a47f68b":{"kind":"REWORK","_number":2,"created":"2022-12-19 08:07:51.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/56/868056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/868056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/868056/2"}}},"commit":{"parents":[{"commit":"2d437e5b4e6e2b1a2a40ef86105c14d43a548883","subject":"feat: support custom render in the TabDetail component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d437e5b4e6e2b1a2a40ef86105c14d43a548883"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-19 06:59:12.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-19 08:07:26.000000000","tz":480},"subject":"fix: fix the delete operation of cluster instance","message":"fix: fix the delete operation of cluster instance\n\nDisable the delete operation if the stack is not ready or the status is `DELETE_IN_PROGRESS`\n\nChange-Id: I18457d3593761e36a7173488b646ebdedf656ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b6e8253a6237e3f5dfc7506fb222ee83a47f68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b6e8253a6237e3f5dfc7506fb222ee83a47f68b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
