)]}'
{"id":"openstack%2Fskyline-console~868155","triplet_id":"openstack%2Fskyline-console~master~I6cb59e9f74c76cb3d0901e89c9bf01d48a85d747","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-20 10:22:51.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-20 11:24:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6cb59e9f74c76cb3d0901e89c9bf01d48a85d747","subject":"fix: fix the magnum quotai and other infos","status":"MERGED","created":"2022-12-20 08:05:48.000000000","updated":"2022-12-20 11:25:51.000000000","submitted":"2022-12-20 11:24:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868155-fix-magnum-bugs","meta_rev_id":"d7b7aac0315e9e4cd704b8b49dedc5586c2d767d","_number":868155,"virtual_id_number":868155,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 11:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-20 10:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-20 10:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 09:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 09:30:14.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"594c90c5b45db07006035fe20f6ecb02bd504a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-20 08:05:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03c3d892d7a46ce75ae083335defe64344e5d16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-20 08:25:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61867fab67c19587b4bd4a9ccbbd62a54e888091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 09:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/813124a5f9e848e8b67641871b4162ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4693a31b684d13b7710fcf843edf50 : SUCCESS in 4m 52s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5d835a94bb7042bb816dcd38757fb4db : SUCCESS in 11m 04s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6204a83e25884281b78b5ca0c6a4d89d : SUCCESS in 6m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b8afb78286844b1a84429261b73bbb20 : SUCCESS in 7m 09s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c4b1250484e347558c1281468766280d : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":2},{"id":"75c72831b6b60131cb6ca76cd774ffddd19fefb1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-20 10:22:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"646c487627dad3b65c70e1f9f293bf31c3dc2756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 10:23:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f957a30aaeb314924e66018a7aae079d70df79b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:24:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9fbbd2aa7941c6b34dcd5b1b148cf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e1ac26ff4944149fb8011368ec89f5 : SUCCESS in 5m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/df24f30efb45456ba14310939b5a001e : SUCCESS in 9m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e2de3eafc7084cd6b27bc30d06fa5002 : SUCCESS in 8m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9959d6e61c494779a13efe31e8de8f21 : SUCCESS in 5m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c7b5fc8b63404eaab358e05e3f320e83 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c406ba78998418f12a07c51ae63043f848fc4f80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:24:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d7b7aac0315e9e4cd704b8b49dedc5586c2d767d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 11:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e3d6a0adbe460994296edbd7d3743c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56e69d4a5fc4466a9a646ae1ae8371f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e03182b072507dfa74182c04ddbd20198af58591","revisions":{"5abf92a213a1d8a3e5727b906d8f2da4d7c6ee68":{"kind":"REWORK","_number":1,"created":"2022-12-20 08:05:48.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/55/868155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/868155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/1"}}},"commit":{"parents":[{"commit":"d48aca89a4289bdea809c625bbe25f9741f3dac0","subject":"fix: reset the subnet list when create cluster instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d48aca89a4289bdea809c625bbe25f9741f3dac0"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-20 08:04:38.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-20 08:05:16.000000000","tz":480},"subject":"fix: fix the magnum quotai and other infos","message":"fix: fix the magnum quotai and other infos\n\n1. support volume quota when create cluster instance\n2. fix subnet reseting when fixed network changed\n3. show project id of cluster template if switch to administrator platform\n\nCloses-Bug: #2000141\nChange-Id: I6cb59e9f74c76cb3d0901e89c9bf01d48a85d747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5abf92a213a1d8a3e5727b906d8f2da4d7c6ee68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5abf92a213a1d8a3e5727b906d8f2da4d7c6ee68"}]},"branch":"refs/heads/master"},"e03182b072507dfa74182c04ddbd20198af58591":{"kind":"REWORK","_number":2,"created":"2022-12-20 08:25:22.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/55/868155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/868155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/868155/2"}}},"commit":{"parents":[{"commit":"d48aca89a4289bdea809c625bbe25f9741f3dac0","subject":"fix: reset the subnet list when create cluster instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d48aca89a4289bdea809c625bbe25f9741f3dac0"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-20 08:04:38.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-20 08:25:07.000000000","tz":480},"subject":"fix: fix the magnum quotai and other infos","message":"fix: fix the magnum quotai and other infos\n\n1. support volume quota when create cluster instance\n2. fix subnet reseting when fixed network changed\n3. show project id of cluster template if switch to administrator platform\n\nCloses-Bug: #2000141\nChange-Id: I6cb59e9f74c76cb3d0901e89c9bf01d48a85d747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e03182b072507dfa74182c04ddbd20198af58591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e03182b072507dfa74182c04ddbd20198af58591"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
