)]}'
{"id":"openstack%2Fskyline-console~868250","triplet_id":"openstack%2Fskyline-console~master~Ib5de517a3c894168c9680a1b37237437a497906f","project":"openstack/skyline-console","branch":"master","topic":"fix-magnum-bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2022-12-21 02:28:29.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2022-12-21 03:23:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5de517a3c894168c9680a1b37237437a497906f","subject":"fix: fix the project column","status":"MERGED","created":"2022-12-21 01:26:19.000000000","updated":"2022-12-21 03:24:45.000000000","submitted":"2022-12-21 03:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868250-fix-magnum-bugs","meta_rev_id":"a0a64df4e1a9a112064398645a3e6c2887b1411e","_number":868250,"virtual_id_number":868250,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-21 03:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2022-12-21 02:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-12-21 02:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 02:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 02:26:36.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b943c5c63a223d58004bb88d8b83d2e2cd8347e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2022-12-21 01:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1abf8321608c166843c3d1630a5688d84552a13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 02:25:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a239cf660344a6f8cc6a6590f54661f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02fd0be431e43f6af0eb5d64ac84d29 : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fcaf2a29a68741af802532fd9f8a5dcb : SUCCESS in 10m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d431d9e8eff84884878383d0b605bcf9 : SUCCESS in 8m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/23e27ff9065447e1b9d20ce597283e97 : SUCCESS in 6m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/de5945bd0b26404486a62fd1fea01d46 : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"12b2fd6ed242690b7004110ac882aa4c521c8799","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-12-21 02:28:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ac7d5378931507a9c71c3af0cbff655c7cad1dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 02:29:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfcbf9967f14b9bb4f8f63b610eebd2dd0a40486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 03:23:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4c3e480181a4705b489c960120e61a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef5e27f254e4414b7491b1f6226932a : SUCCESS in 5m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a1720203a8684581838865d35cbe6c54 : SUCCESS in 9m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d4d6fcf0484e4f26ba32071083e6abed : SUCCESS in 6m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5b92c3a19f9f41a0a6a6d5208230cba2 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/53a974842b60409581e1f1a2aea9dc2e : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":1},{"id":"88faa8e3ef5246c5bd48534a806371dfb1c8e1b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 03:23:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a0a64df4e1a9a112064398645a3e6c2887b1411e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 03:24:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0722667070841e1a931c53f837ee810\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d2afcf243d4ce89f597e91c6f5c625 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"331c2ddd3be1a69ad95bf35c721019cccd9c61d7","revisions":{"331c2ddd3be1a69ad95bf35c721019cccd9c61d7":{"kind":"REWORK","_number":1,"created":"2022-12-21 01:26:19.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/50/868250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/50/868250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/868250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/868250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/50/868250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/50/868250/1"}}},"commit":{"parents":[{"commit":"e03182b072507dfa74182c04ddbd20198af58591","subject":"fix: fix the magnum quotai and other infos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e03182b072507dfa74182c04ddbd20198af58591"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-21 01:24:42.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2022-12-21 01:24:42.000000000","tz":480},"subject":"fix: fix the project column","message":"fix: fix the project column\n\nadd project column to cluster template but not simple flavor column\n\nChange-Id: Ib5de517a3c894168c9680a1b37237437a497906f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/331c2ddd3be1a69ad95bf35c721019cccd9c61d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/331c2ddd3be1a69ad95bf35c721019cccd9c61d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[]}
